)]}'
{"id":"openstack%2Fstevedore~667974","triplet_id":"openstack%2Fstevedore~stable%2Frocky~I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","project":"openstack/stevedore","branch":"stable/rocky","topic":"bindep-fallback","hashtags":[],"change_id":"I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","subject":"Add local bindep.txt and update sphinx logging","status":"MERGED","created":"2019-06-27 16:10:26.000000000","updated":"2019-07-17 12:15:02.000000000","submitted":"2019-07-17 12:15:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667974-1563365702650-77280bbb","meta_rev_id":"e38666497107bb229846bd31bd6d12c0161dac68","_number":667974,"virtual_id_number":667974,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-17 09:01:24.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-17 12:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-10 15:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-16 14:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-07-16 14:21:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 15:09:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-16 14:21:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-17 12:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f342c3805fd7ac78b75debaa5aef462a5fded8a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 16:10:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecb4f595d2bdac27dd776679022339c82a263b4e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 16:45:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d288ecaf3f51c367d220396d26b445a3c566aae","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 17:08:46.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"4a9d0ef92aade2977cfef9567d6341409c7b5a69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 18:27:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/2/check/openstack-tox-pep8/297fca9/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/2/check/openstack-tox-py27/788fd52/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/2/check/openstack-tox-py35/6a97682/ : SUCCESS in 3m 27s\n- requirements-check http://logs.openstack.org/74/667974/2/check/requirements-check/a5f52c0/ : FAILURE in 2m 15s\n- tempest-full http://logs.openstack.org/74/667974/2/check/tempest-full/2dc3931/ : SUCCESS in 1h 34m 47s\n- openstack-tox-docs http://logs.openstack.org/74/667974/2/check/openstack-tox-docs/6cb4d66/ : POST_FAILURE in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/2/check/build-openstack-releasenotes/8396bfe/html/ : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/2/check/openstack-tox-lower-constraints/dc336ae/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a1746fd1ab898b3b0f864473cfff08b3af01f608","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-10 13:40:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e80b0e223c2e577a72711b760c49d71ca5d78005","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-10 14:11:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d8586acb9b166aae3806b90c16c09326d3c3769","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-10 14:29:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca3a2ac8e0f0ed8099781ed26739395e5c0cf859","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-10 14:53:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"188b287534e0e673cc7edaf0b0efa7a3e3aa26eb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-10 15:09:12.000000000","message":"Patch Set 6: Code-Review+2\n\nPity we need to squash these commits but such is life.","accounts_in_message":[],"_revision_number":6},{"id":"45c80c8cba1b5f866ac15dea338bb5f2d4e2cb5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:45:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/6/check/openstack-tox-pep8/a059270/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/6/check/openstack-tox-py27/158a586/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/6/check/openstack-tox-py35/a7e887e/ : SUCCESS in 4m 46s\n- tempest-full http://logs.openstack.org/74/667974/6/check/tempest-full/00b7097/ : SUCCESS in 1h 23m 16s\n- openstack-tox-docs http://logs.openstack.org/74/667974/6/check/openstack-tox-docs/8cd95eb/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/6/check/build-openstack-releasenotes/9f1c4f2/html/ : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/6/check/openstack-tox-lower-constraints/6d4e707/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":6},{"id":"1f2996e69650d3ea1acd792dfd9899d9de4f53c3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-16 14:21:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b30294c07313aa97ccdf8521625449ecac5503fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:21:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b01b887aa1086dd06df806be5bf9bb9546583470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:54:49.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/6/gate/openstack-tox-pep8/fcd3b72/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/6/gate/openstack-tox-py27/450923e/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/6/gate/openstack-tox-py35/5f190e3/ : SUCCESS in 4m 46s\n- tempest-full http://logs.openstack.org/74/667974/6/gate/tempest-full/c4a2e8f/ : FAILURE in 1h 26m 54s\n- openstack-tox-docs http://logs.openstack.org/74/667974/6/gate/openstack-tox-docs/c63a832/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/6/gate/build-openstack-releasenotes/2875303/html/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/6/gate/openstack-tox-lower-constraints/6638d5e/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":6},{"id":"eb832bb6f55ed0429d2f8eca824b62ec1906c53b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-16 16:05:16.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f5aac1bc15867c7fef4e296e3c34ae97c0fcc624","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:28:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/6/check/openstack-tox-pep8/609b7c8/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/6/check/openstack-tox-py27/24b9530/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/6/check/openstack-tox-py35/6c6237f/ : SUCCESS in 3m 09s\n- tempest-full http://logs.openstack.org/74/667974/6/check/tempest-full/1f1f768/ : TIMED_OUT in 2h 11m 56s\n- openstack-tox-docs http://logs.openstack.org/74/667974/6/check/openstack-tox-docs/030dd9e/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/6/check/build-openstack-releasenotes/aeb49c0/html/ : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/6/check/openstack-tox-lower-constraints/d289790/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":6},{"id":"d23ec5a16eb991cc8de6f467ec6e24f83badb4e5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-17 09:01:24.000000000","message":"Patch Set 6:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":6},{"id":"f5a283113028f4151a47d4328ffddbb57a6b599d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 10:35:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/6/check/openstack-tox-pep8/1b40b1e/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/6/check/openstack-tox-py27/5d05d71/ : SUCCESS in 3m 22s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/6/check/openstack-tox-py35/dd80666/ : SUCCESS in 4m 28s\n- tempest-full http://logs.openstack.org/74/667974/6/check/tempest-full/04386f4/ : SUCCESS in 1h 31m 32s\n- openstack-tox-docs http://logs.openstack.org/74/667974/6/check/openstack-tox-docs/045e231/html/ : SUCCESS in 3m 14s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/6/check/build-openstack-releasenotes/6d237d6/html/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/6/check/openstack-tox-lower-constraints/fa6d9d8/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":6},{"id":"65d7306d431e43c06463f8f0b3636044cc7cb35c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 10:35:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"df11be0fefa2909e4a16d6b4e1b1bd4587f01960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 12:15:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"743cc90d33fb29fb067469e3a1e3ad1b26e0a538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 12:15:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/667974/6/gate/openstack-tox-pep8/5f04bb4/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/74/667974/6/gate/openstack-tox-py27/fe8ff50/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/74/667974/6/gate/openstack-tox-py35/555f509/ : SUCCESS in 3m 36s\n- tempest-full http://logs.openstack.org/74/667974/6/gate/tempest-full/c7f9957/ : SUCCESS in 1h 34m 43s\n- openstack-tox-docs http://logs.openstack.org/74/667974/6/gate/openstack-tox-docs/bc239aa/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/74/667974/6/gate/build-openstack-releasenotes/0f84883/html/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/667974/6/gate/openstack-tox-lower-constraints/3eff982/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"39f61e4392ef3c0f1852b87a9d2cb0bc4f410bec","revisions":{"54d7191fb3217ba6777e99d903a7a7cce30c89fc":{"kind":"REWORK","_number":1,"created":"2019-06-27 16:10:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/1"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 16:10:07.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/54d7191fb3217ba6777e99d903a7a7cce30c89fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/54d7191fb3217ba6777e99d903a7a7cce30c89fc"}]},"branch":"refs/heads/stable/rocky"},"1864bfac889ada3efbcbda3308b90db38cf0046a":{"kind":"REWORK","_number":2,"created":"2019-06-27 16:45:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/2"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 16:44:16.000000000","tz":120},"subject":"Add local bindep.txt, update sphinx requirement","message":"Add local bindep.txt, update sphinx requirement\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\nSphinx requirements has to adjust to global requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1864bfac889ada3efbcbda3308b90db38cf0046a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1864bfac889ada3efbcbda3308b90db38cf0046a"}]},"branch":"refs/heads/stable/rocky"},"10cd61fe113c0969a91e762b78aa0227cc668f37":{"kind":"REWORK","_number":3,"created":"2019-07-10 13:40:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/3"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-10 13:39:30.000000000","tz":120},"subject":"Add local bindep.txt, update sphinx requirement","message":"Add local bindep.txt, update sphinx requirement\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\nSphinx requirements has to adjust to global requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/10cd61fe113c0969a91e762b78aa0227cc668f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/10cd61fe113c0969a91e762b78aa0227cc668f37"}]},"branch":"refs/heads/stable/rocky"},"e165848a066abb2c2502b003f99f70e4dc3fd7a4":{"kind":"REWORK","_number":4,"created":"2019-07-10 14:11:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/4"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-10 14:11:01.000000000","tz":120},"subject":"Add local bindep.txt, update sphinx requirement","message":"Add local bindep.txt, update sphinx requirement\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\nSphinx requirements has to adjust to global requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e165848a066abb2c2502b003f99f70e4dc3fd7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e165848a066abb2c2502b003f99f70e4dc3fd7a4"}]},"branch":"refs/heads/stable/rocky"},"a042651936f1a7d2e362c0283f69633b676909dc":{"kind":"REWORK","_number":5,"created":"2019-07-10 14:29:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/5"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-10 14:24:54.000000000","tz":120},"subject":"Add local bindep.txt, update sphinx requirement and sphinx logging","message":"Add local bindep.txt, update sphinx requirement and sphinx logging\n\nThis is a combination of two commits to fix the check / gate jobs:\n\n1. Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n\n2. Update sphinx logging to not use app object\n\nLogging via the application object was deprecated in sphinx 1.6 and is\nnow removed in the master branch that will become 2.0. This updates our\nsphinx extension to use the recommended sphinx.util.logging instead.\n\nCloses-bug: #1798174\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 5b39123aa4d2988fc773003d166c2bde590ea080)\n\nSphinx requirements has to be adjusted to global requirements, too.\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a042651936f1a7d2e362c0283f69633b676909dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a042651936f1a7d2e362c0283f69633b676909dc"}]},"branch":"refs/heads/stable/rocky"},"39f61e4392ef3c0f1852b87a9d2cb0bc4f410bec":{"kind":"REWORK","_number":6,"created":"2019-07-10 14:53:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/667974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/74/667974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/74/667974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/74/667974/6"}}},"commit":{"parents":[{"commit":"1861a85f13074365c3d2eee5e78612e52f710a24","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1861a85f13074365c3d2eee5e78612e52f710a24"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-10 14:52:17.000000000","tz":120},"subject":"Add local bindep.txt and update sphinx logging","message":"Add local bindep.txt and update sphinx logging\n\nThis is a combination of two commits to fix the check / gate jobs:\n\n1. Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n\n2. Update sphinx logging to not use app object\n\nLogging via the application object was deprecated in sphinx 1.6 and is\nnow removed in the master branch that will become 2.0. This updates our\nsphinx extension to use the recommended sphinx.util.logging instead.\n\nCloses-bug: #1798174\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 5b39123aa4d2988fc773003d166c2bde590ea080)\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/39f61e4392ef3c0f1852b87a9d2cb0bc4f410bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/39f61e4392ef3c0f1852b87a9d2cb0bc4f410bec"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
