)]}'
{"id":"openstack%2Fstevedore~667980","triplet_id":"openstack%2Fstevedore~stable%2Fpike~I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","project":"openstack/stevedore","branch":"stable/pike","topic":"bindep-fallback","hashtags":[],"change_id":"I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-27 16:19:14.000000000","updated":"2019-07-19 09:52:49.000000000","submitted":"2019-07-19 09:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667980-1563529969890-77280bbb","meta_rev_id":"897c8ac913639a096851e8cdf39f26515170fc80","_number":667980,"virtual_id_number":667980,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-19 09:52:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-19 08:18:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-17 10:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-19 08:18:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 10:40:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-19 08:18:27.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-19 09:52:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8deca8550de66bf566441d04f94253be4bdac8b7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 16:19:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beabb1b5a07e3c89a056803681764ea1440ec258","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 17:09:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4122406023515008852d2d3814492861b9ada7a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 18:10:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/80/667980/1/check/openstack-tox-pep8/7424e9f/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/80/667980/1/check/openstack-tox-py27/63b3829/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/80/667980/1/check/openstack-tox-py35/19e33d3/ : SUCCESS in 3m 37s\n- tempest-full http://logs.openstack.org/80/667980/1/check/tempest-full/de0c0aa/ : SUCCESS in 1h 28m 16s\n- openstack-tox-docs http://logs.openstack.org/80/667980/1/check/openstack-tox-docs/87b594b/ : POST_FAILURE in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/80/667980/1/check/build-openstack-releasenotes/db6eb31/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"457158d5078169dd1cdae51911d32e2cbf3dc7be","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-10 16:29:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12fc754a6609d4767324cb06eb1060b95b6f1eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:04:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/667980/2/check/openstack-tox-pep8/52ae331/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/80/667980/2/check/openstack-tox-py27/cd29351/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/80/667980/2/check/openstack-tox-py35/9863c1f/ : SUCCESS in 4m 06s\n- requirements-check http://logs.openstack.org/80/667980/2/check/requirements-check/1aae2ba/ : SUCCESS in 2m 27s\n- tempest-full http://logs.openstack.org/80/667980/2/check/tempest-full/5cd5326/ : SUCCESS in 1h 28m 50s\n- openstack-tox-docs http://logs.openstack.org/80/667980/2/check/openstack-tox-docs/2b163ea/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/80/667980/2/check/build-openstack-releasenotes/f1b237b/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1b024e01542278b0f8c06f78ff367f73edcf3c21","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-17 10:40:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b75373a9ca7f808a0e1e78bbe3ed5390de889895","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-19 08:18:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"de1167b45de89f2a9db3d652ffb7901e34eca8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 08:18:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e1da360de7270981f8b7b4fc6524ff9d7ae4671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 09:52:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8ee3026262c13c52ce5973bf09228f17409fcbdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 09:52:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/667980/2/gate/openstack-tox-pep8/9a5ee9c/ : SUCCESS in 2m 29s\n- openstack-tox-py27 http://logs.openstack.org/80/667980/2/gate/openstack-tox-py27/17c88b6/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/80/667980/2/gate/openstack-tox-py35/9485e00/ : SUCCESS in 3m 22s\n- requirements-check http://logs.openstack.org/80/667980/2/gate/requirements-check/67bd0a3/ : SUCCESS in 2m 13s\n- tempest-full http://logs.openstack.org/80/667980/2/gate/tempest-full/24fbc77/ : SUCCESS in 1h 33m 12s\n- openstack-tox-docs http://logs.openstack.org/80/667980/2/gate/openstack-tox-docs/07bb494/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/80/667980/2/gate/build-openstack-releasenotes/c10f495/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e00bdabb80eabd13251b7f5b578b0cee231b35ae","revisions":{"405a66a66f62a8658a4ac4ddd2555eaebf6a6d99":{"kind":"REWORK","_number":1,"created":"2019-06-27 16:19:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/667980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/80/667980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/80/667980/1"}}},"commit":{"parents":[{"commit":"66662c9eeb172f7d1aaec713d050029d1c37ce7e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/66662c9eeb172f7d1aaec713d050029d1c37ce7e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 16:18:52.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n(cherry picked from commit 54d7191fb3217ba6777e99d903a7a7cce30c89fc)\n(cherry picked from commit 78ffc8d4b0702f30d73f49e5f265cc1c371b5b40)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/405a66a66f62a8658a4ac4ddd2555eaebf6a6d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/405a66a66f62a8658a4ac4ddd2555eaebf6a6d99"}]},"branch":"refs/heads/stable/pike"},"e00bdabb80eabd13251b7f5b578b0cee231b35ae":{"kind":"REWORK","_number":2,"created":"2019-07-10 16:29:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/667980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/80/667980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/80/667980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/80/667980/2"}}},"commit":{"parents":[{"commit":"66662c9eeb172f7d1aaec713d050029d1c37ce7e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/66662c9eeb172f7d1aaec713d050029d1c37ce7e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-10 16:28:21.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n(cherry picked from commit caf65ed26066b5f7d13cc50e67f1df852414b2dc)\n\nConflicts:\n      doc/requirements.txt\n      test-requirements.txt\nNote(elod.illes): conflict is due to not having proposal bot\u0027s updates\nfrom global requirements.\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit 39f61e4392ef3c0f1852b87a9d2cb0bc4f410bec)\n(cherry picked from commit f10075c4b36cd2375234e8bb93247be13869332a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e00bdabb80eabd13251b7f5b578b0cee231b35ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e00bdabb80eabd13251b7f5b578b0cee231b35ae"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
