)]}'
{"id":"openstack%2Fstevedore~669469","triplet_id":"openstack%2Fstevedore~master~Idcfef4d6661e51ba36caa0885e36df3784e483ab","project":"openstack/stevedore","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Idcfef4d6661e51ba36caa0885e36df3784e483ab","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:15:25.000000000","updated":"2019-07-05 23:02:41.000000000","submitted":"2019-07-05 23:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669469-1562367761593-29425bd4","meta_rev_id":"38ec6c85627e90a32bff94811dd3f88ea52f925e","_number":669469,"virtual_id_number":669469,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-07-05 23:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":2,"date":"2019-07-05 19:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":1,"date":"2019-07-05 19:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 19:28:39.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-07-05 23:02:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"094daffaf0094507ff2f9b3fd064ff4ac3ab6f20","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10bbbc72f022b956216c80e9831acc499636b7c1","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-07-05 19:28:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb030465a538dee5666100e8b8ea0080f6aad760","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:12:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/669469/1/check/test-release-openstack/a698c1f/ : SUCCESS in 2m 26s\n- tempest-full http://logs.openstack.org/69/669469/1/check/tempest-full/b0ea2e5/ : SUCCESS in 1h 50m 48s\n- tempest-full-py3 http://logs.openstack.org/69/669469/1/check/tempest-full-py3/7ab60d0/ : SUCCESS in 1h 47m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/669469/1/check/openstack-tox-lower-constraints/150d415/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/69/669469/1/check/openstack-tox-pep8/3554814/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/69/669469/1/check/openstack-tox-py27/62a0335/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/69/669469/1/check/openstack-tox-py36/2ef3552/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/69/669469/1/check/openstack-tox-py37/24af5ab/ : SUCCESS in 6m 49s\n- openstack-tox-docs http://logs.openstack.org/69/669469/1/check/openstack-tox-docs/2bee37e/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/69/669469/1/check/build-openstack-releasenotes/d5a060b/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f1b2cbf0564bba60b83e802537f902cfdc56bdcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:12:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15bf5aebe41ce62f03b040084c3d98f6c90b763e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 23:02:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca99dc96b2f5935fa91cf5e8e16618a22c8cba6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 23:02:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/69/669469/1/gate/test-release-openstack/2e9be96/ : SUCCESS in 2m 30s\n- tempest-full http://logs.openstack.org/69/669469/1/gate/tempest-full/5ab3655/ : SUCCESS in 1h 48m 53s\n- tempest-full-py3 http://logs.openstack.org/69/669469/1/gate/tempest-full-py3/c2665c8/ : SUCCESS in 1h 33m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/669469/1/gate/openstack-tox-lower-constraints/6392ac0/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/69/669469/1/gate/openstack-tox-pep8/2892d6d/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/69/669469/1/gate/openstack-tox-py27/a706de6/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/69/669469/1/gate/openstack-tox-py36/8072384/ : SUCCESS in 2m 38s\n- openstack-tox-py37 http://logs.openstack.org/69/669469/1/gate/openstack-tox-py37/1b99d89/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/69/669469/1/gate/openstack-tox-docs/54a5b2d/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/69/669469/1/gate/build-openstack-releasenotes/b022cd7/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6817543c3003a29f65729703f63ed3d35d30ff84","revisions":{"6817543c3003a29f65729703f63ed3d35d30ff84":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:15:25.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/69/669469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/69/669469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/669469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/669469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/669469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/69/669469/1"}}},"commit":{"parents":[{"commit":"feac47186d79569bb564e18216e07f18ed5ccd8b","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/feac47186d79569bb564e18216e07f18ed5ccd8b"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:57:00.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:15:24.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Idcfef4d6661e51ba36caa0885e36df3784e483ab\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/6817543c3003a29f65729703f63ed3d35d30ff84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/6817543c3003a29f65729703f63ed3d35d30ff84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
