)]}'
{"id":"openstack%2Fstevedore~687801","triplet_id":"openstack%2Fstevedore~master~I0573019ba9e98d31ccb43179590a970c9b7647af","project":"openstack/stevedore","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I0573019ba9e98d31ccb43179590a970c9b7647af","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-10-10 07:05:32.000000000","updated":"2019-12-19 09:33:09.000000000","submitted":"2019-12-19 09:30:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687801-1576747829164-0d4ca9dc","meta_rev_id":"8713f5429dba14bd39a290d33715376e93b28cef","_number":687801,"virtual_id_number":687801,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 09:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-12-19 09:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-10-28 15:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-19 09:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 15:27:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-19 09:18:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-19 09:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34889229366b0b5bf57b170bf880f5d208338942","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-10 07:05:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4d7a5f55fb6d169d2808936462cd29cc49efb28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 07:10:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/80b91a09403a42bb9bceaa2cea2aea77 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8026073140468b822c8300d4fdc6b5 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12be4c3ec52b43978fce091e786b0bb7 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5840aa4c2fe749ba82656d04e2c3cdda : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"89445f86c94d905f33ff0530b82197a1d632d054","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-28 15:27:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"336bbfe14c7b48afd5ec16cc5b167fac11598dbb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-19 09:18:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSure, why not","accounts_in_message":[],"_revision_number":1},{"id":"ca1af7d7c63309108cb747a65c570e56f5596e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ddd5eb25063f1c7365732355dedacd8c01b9654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:30:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/03972511e6174c8cbafb98cc39e411e8 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed66b79fe1a43fbb3d3a163763f5f9e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5bfc2d5c87457dbcb203d2f33f4a03 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8061fb53642a45e499860778dd7c6ca7 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"01e68185d9e72bff98e00a24eb8ec13de19d2592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:30:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c42478aea49237752c042f42a9174e75fb001b15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:33:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0919051e2e24365993fcd0e4fb149d9 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d7334552f464d7eb08400e7bd8d55d6 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22d92824d2b94017d4f53befa881130d272e34e9","revisions":{"22d92824d2b94017d4f53befa881130d272e34e9":{"kind":"REWORK","_number":1,"created":"2019-10-10 07:05:32.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/01/687801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/01/687801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/01/687801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/01/687801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/01/687801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/01/687801/1"}}},"commit":{"parents":[{"commit":"234a272d9884a38ccd34563ae7a7afd03dfd54c9","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/234a272d9884a38ccd34563ae7a7afd03dfd54c9"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-10 07:05:44.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-10 07:05:44.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: I0573019ba9e98d31ccb43179590a970c9b7647af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/22d92824d2b94017d4f53befa881130d272e34e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/22d92824d2b94017d4f53befa881130d272e34e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
