)]}'
{"id":"openstack%2Fstevedore~739306","triplet_id":"openstack%2Fstevedore~master~I8b08f289d446f4775eac1e1a91997fa96f25f641","project":"openstack/stevedore","branch":"master","topic":"osc-performance","hashtags":[],"change_id":"I8b08f289d446f4775eac1e1a91997fa96f25f641","subject":"switch to importlib.metadata package","status":"MERGED","created":"2020-07-04 16:09:32.000000000","updated":"2020-07-10 17:00:29.000000000","submitted":"2020-07-10 16:57:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739306-1594400233173-5d897941","meta_rev_id":"70bec423b0dbe4172147d3d40e29b83f69333cb6","_number":739306,"virtual_id_number":739306,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-07-06 18:15:20.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 16:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 15:36:44.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-07-06 17:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-07-06 16:08:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-10 15:36:44.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 20:06:46.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-06 16:08:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-06 17:30:59.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-10 15:36:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 16:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e5b0b94f3e78686e878b46892406e85866d494c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 16:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba98bf5e7eb6f7e1a24206f1767f14da11f4ca0e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 16:10:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"144164e6c2d9489aac6af522ecee0df7a93d226c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 16:25:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"991be3af9747636d233d99dbfb54745e46dfb15b","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 16:25:50.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"eef0d6df9a02b620365aeb6340a7193265a8e9df","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 16:27:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fdde269e5250600eed1a561cce73d5e392b42ab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 17:21:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83b3dfd7345648fcb6b836f9397b9561 : SUCCESS in 53m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0765fb730433434390a94bed4a0384ed : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a0f267b9fd246c38fb59f21e91001d5 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6ce46a16e16497b9d3dbb61992ae64d : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab0e7de4c01441fa38824324c22102e : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60ff2734e3742f487bc9007e42e5b8a : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"6f8b5394053a01d9a07801d64d5466cd6cc2a09f","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 19:19:01.000000000","message":"Patch Set 3: -Code-Review\n\nThis uses a vendored copy of the \u0027entrypoints\u0027 project, for now. The public API for the cache is added in the follow-up patch. If a new release of entrypoints is made, we could drop the vendored copy here and retain the public API.","accounts_in_message":[],"_revision_number":3},{"id":"2b6c2f162bfd83cd04470d42fee5efce5f26c1f8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-04 20:20:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"019dc93ab4ce48e1f6a38d434c267fead0e6fa77","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 19:18:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0d20c8013e5e888bbe4a35afd1b6e3c38c07dc69","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 19:35:09.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"cfd4aca3023f50d0bd4f590f23193b8eae1ef234","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 20:05:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"572a07391d5e39c8e467bddc032874e2ea1cee1e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 20:06:30.000000000","message":"Patch Set 6:\n\nThis change will not be 100% backwards compatible if anything was relying on the use of pkg_resources.EntryPoint. The APIs of stevedore are otherwise still compatible, but we should probably increment the major version # on the next release to be clear.","accounts_in_message":[],"_revision_number":6},{"id":"a9ce89d1aa9fe4b4d1341c078fc999bc5899f2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 21:29:05.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf88c14473534e04aebb78bada42824e : FAILURE in 2m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90128f0a29884023ab10574339aace2c : SUCCESS in 1h 22m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a607abfd0d94078992cf4d1cfeef98f : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb43f782dad444ea206082b65fc5274 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32d7c6587b074f06a60b86ba79cb8bc9 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/825139deaa1646a98f4c7bbb9c2fb490 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2916c52932354405a6c051f3b056083e : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":6},{"id":"8e1d855adbf70ea3340aa1c4c3c390c80ce4340b","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 22:14:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"afd25912405e0867bcecacd394db9548f160075a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 23:32:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/52808826c0a44babbe64c04adf5c1c3f : FAILURE in 4m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d9bdab4256f4fc1869acca584ca1133 : SUCCESS in 1h 17m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c28f60297aa144558654204593f95c9f : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f9513f27ebb4538be0ea4021a28a84a : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8a7454ab4414dd3868e9e175e5dc74e : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b340621c2f5e424e9ac1c6158a57a6ca : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0ce52d425641b8927cede418c0b451 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":7},{"id":"3c566043059ccf79fa2ecac85651aa0e4775f4f7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 23:39:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4d423cac1fe3757dede601f6aaadee02bd60c324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 01:02:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/83a258dcf35e4113922900603b96aa44 : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eef1e2f8b64c4b179e332bc11e215eb7 : SUCCESS in 1h 20m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e01ec9f3382e4f5e8f4650148289fa50 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e590b729632144bcb9fd4ffe62878214 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d29ecd458c444b56ba7b61d6dae95cbc : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7d733e0831a4b5c87e51a6dc6a5d656 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22aa3599f36462189eb01a0965d04e1 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":8},{"id":"26609892ac6dea6878b592c9237aa2a81e5860c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 08:30:35.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"883c3f2c8737ee6c8bd5eb9efb5458525ee7fd1a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 08:43:08.000000000","message":"Patch Set 8:\n\nHello,\n\nNice changes, especially the caching utility.\n\n \u003e This change will not be 100% backwards compatible if anything was\n \u003e relying on the use of pkg_resources.EntryPoint. The APIs of\n \u003e stevedore are otherwise still compatible, but we should probably\n \u003e increment the major version # on the next release to be clear.\n\nDo we need these changes on stable branches? If yes I think we could split these changes to allow us to only backport the library switching  and to extract the caching utility to avoid to backport a new feature on stable branches.\n\nSemantically these changes are loosely coupled, so I think the caching part could be submitted via a follow up patch.\n\nThoughts?","accounts_in_message":[],"_revision_number":8},{"id":"c07114e45875b13108292d018cdbcc951122bf81","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 09:40:34.000000000","message":"Patch Set 8:\n\n(my -1 is only related to the missing header)","accounts_in_message":[],"_revision_number":8},{"id":"1320ecfe09dd5be8355a89ba8c088334b4314768","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-06 12:40:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5ec7a266d3dafbe55aedc52ade1d75baa7708c64","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-06 12:41:25.000000000","message":"Patch Set 8:\n\n\u003e Hello,\n \u003e \n \u003e Nice changes, especially the caching utility.\n \u003e \n \u003e \u003e This change will not be 100% backwards compatible if anything was\n \u003e \u003e relying on the use of pkg_resources.EntryPoint. The APIs of\n \u003e \u003e stevedore are otherwise still compatible, but we should probably\n \u003e \u003e increment the major version # on the next release to be clear.\n \u003e \n \u003e Do we need these changes on stable branches? If yes I think we\n \u003e could split these changes to allow us to only backport the library\n \u003e switching  and to extract the caching utility to avoid to backport\n \u003e a new feature on stable branches.\n \u003e\n \u003e Semantically these changes are loosely coupled, so I think the\n \u003e caching part could be submitted via a follow up patch.\n \u003e \n \u003e Thoughts?\n\n\nI don\u0027t think we need to make the change on a stable branch. The cache is a new feature, and since it results in an incompatibility I don\u0027t think we want to backport it.\n\nI fixed the header issue. Thanks for pointing that out, I didn\u0027t think of it and I guess we don\u0027t have a linter rule for that.","accounts_in_message":[],"_revision_number":8},{"id":"8568a0fe467bdfea69d98174fba72266d08d76e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:16:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a632d8300ec40fe84b10c4fff40b05a : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/145049a37e0a4355bc07e8ff58faae84 : SUCCESS in 1h 28m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42fa0540c50e4457ba5a2930609bcabd : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24878fcba983411bb54a58431704bbe7 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aa0d451f0f3439c86472d78d149b3fc : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e771fa20ddd482ba9b59463efcdbe38 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db64034cd73841a6aef4f15556df10b8 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":9},{"id":"8c9e32992c0655c90360588eb2deea3f84c4f713","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:08:11.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks for the changes :)\n\n \u003e \u003e Hello,\n \u003e \u003e\n \u003e \u003e Nice changes, especially the caching utility.\n \u003e \u003e\n \u003e \u003e \u003e This change will not be 100% backwards compatible if anything\n \u003e was\n \u003e \u003e \u003e relying on the use of pkg_resources.EntryPoint. The APIs of\n \u003e \u003e \u003e stevedore are otherwise still compatible, but we should\n \u003e probably\n \u003e \u003e \u003e increment the major version # on the next release to be clear.\n \u003e \u003e\n \u003e \u003e Do we need these changes on stable branches? If yes I think we\n \u003e \u003e could split these changes to allow us to only backport the\n \u003e library\n \u003e \u003e switching  and to extract the caching utility to avoid to\n \u003e backport\n \u003e \u003e a new feature on stable branches.\n \u003e \u003e\n \u003e \u003e Semantically these changes are loosely coupled, so I think the\n \u003e \u003e caching part could be submitted via a follow up patch.\n \u003e \u003e\n \u003e \u003e Thoughts?\n \u003e \n \u003e \n \u003e I don\u0027t think we need to make the change on a stable branch. The\n \u003e cache is a new feature, and since it results in an incompatibility\n \u003e I don\u0027t think we want to backport it.\n\nThis is that I was thinking too but I prefered ask about this to be sure.\n \u003e \n \u003e I fixed the header issue. Thanks for pointing that out, I didn\u0027t\n \u003e think of it and I guess we don\u0027t have a linter rule for that.\n\nYes we don\u0027t have a linter rule for that. In hacking we have a rule dedicated to check if licence used is the apache licence [1], but AFAIK it has nothing related to missing licence header. I think a rule can be added to hacking and to implement it we need to possibly ignore `__init__.py` (if file is empty).\n\n[1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L75","accounts_in_message":[],"_revision_number":9},{"id":"7c1b5df8cf71b3257d61884cbcc4202eb7be0dc2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:10:59.000000000","message":"Patch Set 9:\n\n\u003e Thanks for the changes :)\n \u003e \n \u003e \u003e \u003e Hello,\n \u003e \u003e \u003e\n \u003e \u003e \u003e Nice changes, especially the caching utility.\n \u003e \u003e \u003e\n \u003e \u003e \u003e \u003e This change will not be 100% backwards compatible if anything\n \u003e \u003e was\n \u003e \u003e \u003e \u003e relying on the use of pkg_resources.EntryPoint. The APIs of\n \u003e \u003e \u003e \u003e stevedore are otherwise still compatible, but we should\n \u003e \u003e probably\n \u003e \u003e \u003e \u003e increment the major version # on the next release to be\n \u003e clear.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Do we need these changes on stable branches? If yes I think we\n \u003e \u003e \u003e could split these changes to allow us to only backport the\n \u003e \u003e library\n \u003e \u003e \u003e switching  and to extract the caching utility to avoid to\n \u003e \u003e backport\n \u003e \u003e \u003e a new feature on stable branches.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Semantically these changes are loosely coupled, so I think the\n \u003e \u003e \u003e caching part could be submitted via a follow up patch.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Thoughts?\n \u003e \u003e\n \u003e \u003e\n \u003e \u003e I don\u0027t think we need to make the change on a stable branch. The\n \u003e \u003e cache is a new feature, and since it results in an\n \u003e incompatibility\n \u003e \u003e I don\u0027t think we want to backport it.\n \u003e \n \u003e This is that I was thinking too but I prefered ask about this to be\n \u003e sure.\n \u003e \u003e\n \u003e \u003e I fixed the header issue. Thanks for pointing that out, I didn\u0027t\n \u003e \u003e think of it and I guess we don\u0027t have a linter rule for that.\n \u003e \n \u003e Yes we don\u0027t have a linter rule for that. In hacking we have a rule\n \u003e dedicated to check if licence used is the apache licence [1], but\n \u003e AFAIK it has nothing related to missing licence header. I think a\n \u003e rule can be added to hacking and to implement it we need to\n \u003e possibly ignore `__init__.py` (if file is empty).\n \u003e \n \u003e [1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L75\n\nOh wait I missed that one [1], maybe this one could be fixed to ensure to catch missing header.\n\n[1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L50","accounts_in_message":[],"_revision_number":9},{"id":"008d99096d83d706e1644e59026dc0ed47b2892e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-06 17:30:59.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"86a2bf5d13c0e46ed1ac4327ddbb7c0ba9acb681","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-06 18:15:20.000000000","message":"Patch Set 9:\n\n\u003e \u003e \u003e I fixed the header issue. Thanks for pointing that out, I\n \u003e didn\u0027t\n \u003e \u003e \u003e think of it and I guess we don\u0027t have a linter rule for that.\n \u003e \u003e\n \u003e \u003e Yes we don\u0027t have a linter rule for that. In hacking we have a\n \u003e rule\n \u003e \u003e dedicated to check if licence used is the apache licence [1], but\n \u003e \u003e AFAIK it has nothing related to missing licence header. I think a\n \u003e \u003e rule can be added to hacking and to implement it we need to\n \u003e \u003e possibly ignore `__init__.py` (if file is empty).\n \u003e \u003e\n \u003e \u003e [1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L75\n \u003e \n \u003e Oh wait I missed that one [1], maybe this one could be fixed to\n \u003e ensure to catch missing header.\n \u003e \n \u003e [1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L50\n\nDo we even run hacking against this repo? It isn\u0027t used everywhere.","accounts_in_message":[],"_revision_number":9},{"id":"3658b885267b7634a7a56fc2a68bf6d04460178f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:26:23.000000000","message":"Patch Set 9:\n\n\u003e \u003e \u003e \u003e I fixed the header issue. Thanks for pointing that out, I\n \u003e \u003e didn\u0027t\n \u003e \u003e \u003e \u003e think of it and I guess we don\u0027t have a linter rule for that.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Yes we don\u0027t have a linter rule for that. In hacking we have a\n \u003e \u003e rule\n \u003e \u003e \u003e dedicated to check if licence used is the apache licence [1],\n \u003e but\n \u003e \u003e \u003e AFAIK it has nothing related to missing licence header. I think\n \u003e a\n \u003e \u003e \u003e rule can be added to hacking and to implement it we need to\n \u003e \u003e \u003e possibly ignore `__init__.py` (if file is empty).\n \u003e \u003e \u003e\n \u003e \u003e \u003e [1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L75\n \u003e \u003e\n \u003e \u003e Oh wait I missed that one [1], maybe this one could be fixed to\n \u003e \u003e ensure to catch missing header.\n \u003e \u003e\n \u003e \u003e [1] https://github.com/openstack/hacking/blob/b6bca99c62570a534cb6b7720ab81fd354f2f38e/hacking/checks/comments.py#L50\n \u003e \n \u003e Do we even run hacking against this repo? It isn\u0027t used everywhere.\n\nGood question, I don\u0027t find it in this repo, I will add it via another patch.","accounts_in_message":[],"_revision_number":9},{"id":"589111e68cb4e289570978901aa25d284909a664","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:36:44.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"77c5a5ae07ed19a0f3c63cdc4ede5a51624021bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:36:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"233f83996b73120a89e401deb88c1915357a67f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:57:12.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/994842a6ac0e413b836b7388ec6777ea : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93105a12965f4bd88246f145497a09ae : SUCCESS in 1h 12m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44eb969e27eb4df9b3bea0a1745788fe : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c06421775924c2f8a65d3a5b461c4c0 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0695d266ca1740afba918c854d95cd69 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8183dc8ba744071823a7c794eb2f2e2 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d985380e0b604b9b8f4276cbf73a0122 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":9},{"id":"5119b8575c2753fb0c4fe77dd4be7047d66997f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:57:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"3fb019ec0484014af12b588b95ce5b7f13b78d1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:00:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50349cc11e945a2a0767910cbb577ac : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"d5297167e08468c75d2477f15004df61cf98e57e","revisions":{"42b19b682b0a5068a88571bc235435d957578ea3":{"kind":"REWORK","_number":1,"created":"2020-07-04 16:09:32.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/1"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"subject":"switch to entrypoints package","message":"switch to entrypoints package\n\nLoad entry points using \u0027entrypoints\u0027 instead of \u0027pkg_resources\u0027.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/42b19b682b0a5068a88571bc235435d957578ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/42b19b682b0a5068a88571bc235435d957578ea3"}]},"branch":"refs/heads/master"},"26dfa13c8c337c11af29f285bb7284ec43769b91":{"kind":"REWORK","_number":2,"created":"2020-07-04 16:25:32.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/2"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:23:12.000000000","tz":-240},"subject":"switch to entrypoints package","message":"switch to entrypoints package\n\nLoad entry points using \u0027entrypoints\u0027 instead of \u0027pkg_resources\u0027.\n\nUse a vendored copy of\nhttps://github.com/takluyver/entrypoints/blob/master/entrypoints.py\nfrom\nhttps://github.com/takluyver/entrypoints/commit/bd8d41a33bfa2df7dbf5fddf7f6f2e5bf6b6d95a\nwhich contains an unreleased fix for handling editable distributions\nwhich avoids reporting the same plugins twice when testing under tox.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/26dfa13c8c337c11af29f285bb7284ec43769b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/26dfa13c8c337c11af29f285bb7284ec43769b91"}]},"branch":"refs/heads/master"},"3f5fd24dd55077bbacfcba82a238a03d5b770a0c":{"kind":"REWORK","_number":3,"created":"2020-07-04 16:27:37.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/3"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:27:26.000000000","tz":-240},"subject":"switch to entrypoints package","message":"switch to entrypoints package\n\nLoad entry points using \u0027entrypoints\u0027 instead of \u0027pkg_resources\u0027.\n\nUse a vendored copy of\nhttps://github.com/takluyver/entrypoints/blob/master/entrypoints.py\nfrom\nhttps://github.com/takluyver/entrypoints/commit/bd8d41a33bfa2df7dbf5fddf7f6f2e5bf6b6d95a\nwhich contains an unreleased fix for handling editable distributions\nwhich avoids reporting the same plugins twice when testing under tox.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3f5fd24dd55077bbacfcba82a238a03d5b770a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3f5fd24dd55077bbacfcba82a238a03d5b770a0c"}]},"branch":"refs/heads/master"},"8dd04c8ebd924643e7df0fde3cbdf0e67e10407f":{"kind":"REWORK","_number":4,"created":"2020-07-05 19:18:14.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/4"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 19:17:59.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/8dd04c8ebd924643e7df0fde3cbdf0e67e10407f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/8dd04c8ebd924643e7df0fde3cbdf0e67e10407f"}]},"branch":"refs/heads/master"},"4d3c045174156b5ad6d27858d75e1eb5b4e59c94":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-07-05 19:35:09.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/5"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 19:35:00.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4d3c045174156b5ad6d27858d75e1eb5b4e59c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4d3c045174156b5ad6d27858d75e1eb5b4e59c94"}]},"branch":"refs/heads/master"},"1c58a2e4c36b8986dd3f0c054a6d164b27ffff89":{"kind":"REWORK","_number":6,"created":"2020-07-05 20:05:27.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/6"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 20:05:21.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1c58a2e4c36b8986dd3f0c054a6d164b27ffff89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1c58a2e4c36b8986dd3f0c054a6d164b27ffff89"}]},"branch":"refs/heads/master"},"3375f8c6610036af385ab404c5d94395c90b56fd":{"kind":"REWORK","_number":7,"created":"2020-07-05 22:14:09.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/7"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 22:13:52.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3375f8c6610036af385ab404c5d94395c90b56fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3375f8c6610036af385ab404c5d94395c90b56fd"}]},"branch":"refs/heads/master"},"a5c6db941cf4214e289501c7f2a4918222377e87":{"kind":"REWORK","_number":8,"created":"2020-07-05 23:39:40.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/8"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 23:39:35.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a5c6db941cf4214e289501c7f2a4918222377e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a5c6db941cf4214e289501c7f2a4918222377e87"}]},"branch":"refs/heads/master"},"d5297167e08468c75d2477f15004df61cf98e57e":{"kind":"REWORK","_number":9,"created":"2020-07-06 12:40:05.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/06/739306/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/06/739306/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/06/739306/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/06/739306/9"}}},"commit":{"parents":[{"commit":"01c12eca14dc6a5078b4308a5fc6c3cbef930b06","subject":"Remove Travis CI config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/01c12eca14dc6a5078b4308a5fc6c3cbef930b06"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 16:08:57.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-06 12:39:52.000000000","tz":-240},"subject":"switch to importlib.metadata package","message":"switch to importlib.metadata package\n\nLoad entry points using \u0027importlib.metadata\u0027 instead of \u0027pkg_resources\u0027.\n\nInclude a caching layer.  The cache stores the parsed text data from all\nof the ini input files in a single JSON file with a name based on the\nhash of the path entries and the mtimes. This should produce a unique\nfilename for each import path, regardless of the use of a virtualenv.\n\nThe data is stored in a format that means no other files need to be\nexamined or parsed in order to return EntryPoint objects.\n\nChange-Id: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d5297167e08468c75d2477f15004df61cf98e57e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d5297167e08468c75d2477f15004df61cf98e57e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
