)]}'
{"id":"openstack%2Fstevedore~740595","triplet_id":"openstack%2Fstevedore~master~I42448c2089b5c20f8cfe5d31dc26744db76f6d79","project":"openstack/stevedore","branch":"master","topic":"importlib-metadata","hashtags":[],"change_id":"I42448c2089b5c20f8cfe5d31dc26744db76f6d79","subject":"Update sphinxext to work with different EntryPoints","status":"ABANDONED","created":"2020-07-12 16:03:12.000000000","updated":"2020-07-13 12:47:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdc6b8b57c3fa6028f05c759dad6f1434a9fd45e","_number":740595,"virtual_id_number":740595,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-12 17:19:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":-1,"date":"2020-07-12 21:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-07-12 16:03:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-12 16:03:52.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-12 17:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-12 21:54:56.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"}],"messages":[{"id":"557c85e111e845cd2980b1476135c94c8080be7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-12 16:03:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"368343d6f951ffd1695162f294ec890c16e8ea79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 17:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da65610292474bdfaf217033a0619045 : SUCCESS in 1h 15m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7844d951d0c4aa882b3c80abf4ba856 : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48bae9d3fe8a49f88f9daf93ba6e45be : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c534db674947dbb98130e8b12cb459 : FAILURE in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe1526a2f8804c1da8cd1b915caba1b4 : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec7f8278e0545149718dc7506286a78 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"725feaa5a083b6e13daa0c46910fb14e0e1c99ef","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-12 21:54:56.000000000","message":"Patch Set 1: Code-Review-1\n\nSee https://review.opendev.org/#/c/740596/2 as an alternative","accounts_in_message":[],"_revision_number":1},{"id":"dd1d3f0fd614db92d04b609482d1484b48df00c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-13 12:47:57.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/740596/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d91661f75dc278b412f6624c2261414da6bf1e90","revisions":{"d91661f75dc278b412f6624c2261414da6bf1e90":{"kind":"REWORK","_number":1,"created":"2020-07-12 16:03:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/95/740595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/740595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/740595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/740595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/740595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/740595/1"}}},"commit":{"parents":[{"commit":"5eb3ef2de3576c8633dddab832bbe85405930db0","subject":"add release note before major version update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/5eb3ef2de3576c8633dddab832bbe85405930db0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-12 15:58:07.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-12 15:58:07.000000000","tz":-300},"subject":"Update sphinxext to work with different EntryPoints","message":"Update sphinxext to work with different EntryPoints\n\nWe switched to using importlib.metadata to get EntryPoint information\nin I8b08f289d446f4775eac1e1a91997fa96f25f641, and the sphinx extension\nwas updated to handle the EntryPoints returned by that module.\n\nIt appears that projects that use a mix of projects using\nimportlib.metadata and pkg_resources end up getting a mix of EntryPoint\nobject types. This was noted in Glance doc builds where this extension\nwould error due to an EntryPoint not having the \"module\" property\nexpected while parsing some taskflow related information.\n\nThis adds a helper method to make extracting the module information\nmore robust. It wasn\u0027t enough to just assume one if the other was not\nfound, so the helper will try a variety of values before defaulting\nthe the EntryPoint name.\n\nChange-Id: I42448c2089b5c20f8cfe5d31dc26744db76f6d79\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d91661f75dc278b412f6624c2261414da6bf1e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d91661f75dc278b412f6624c2261414da6bf1e90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
