)]}'
{"id":"openstack%2Fstevedore~742153","triplet_id":"openstack%2Fstevedore~master~I045fa1cd7932d960e1cf49b2c335c7acdeba0b46","project":"openstack/stevedore","branch":"master","topic":"oslo-pre-commit","hashtags":[],"change_id":"I045fa1cd7932d960e1cf49b2c335c7acdeba0b46","subject":"Adding pre-commit","status":"MERGED","created":"2020-07-21 11:40:49.000000000","updated":"2020-10-05 17:01:55.000000000","submitted":"2020-10-05 16:59:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742153-1601917160816-3cd9c318","meta_rev_id":"069263dadf29afbeafd7d66c072423689c086ad5","_number":742153,"virtual_id_number":742153,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-09-22 15:14:03.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 16:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-09-23 08:06:59.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2020-09-28 15:18:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-05 15:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-10-05 15:43:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 08:06:59.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-28 15:18:44.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-05 15:43:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-10-05 16:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25aa37b97c32beec4dee163bdc9a913f0a998d9c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 11:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d4d1ed3564914c923f63ed65a7f5472aaec89a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:02:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/526a21eadab9446f8f2f1c79a6dea818 : SUCCESS in 1h 19m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63eede615ba144e5866d3d14525ce498 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ceb523ab1d464aab9645a51841d124 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3652f8a732a842ad84f2484f36d023aa : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b846de888344c4485913369f82d7d4b : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064794a7cced4b0abf182292c925b27c : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/460af830631d45c5a2eb213089ace609 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9c73f649e86c238e073d83d2880f499ce9a793f7","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:41:41.000000000","message":"Patch Set 1: Code-Review-2\n\nWe should hold on in these patches for a while and observe oslo.cache for some time before spreading the change around all Oslo.","accounts_in_message":[],"_revision_number":1},{"id":"c240d0fdd03680a1f0043b5aebe63a0deb58913b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 11:41:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0368cf8983b290542da1efc8a821c4bb5994df08","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-15 12:59:10.000000000","message":"Removed Code-Review-2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"11ed3186dd7ce8005371ded780825a056763c259","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-15 13:03:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de55afdb243c6e4e42a87383b98776cd89b99906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:08:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/20a410db7ed04624bd0e0423f0924478 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a659997d5a624465bb5b7893bf9ab234 : SUCCESS in 1h 21m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60e8b042aaf34f8eae13a63484fe4197 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b76cc2fbf84736989737dace45ad48 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b7db1b077e6455eb10ce3eb5911c7a4 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78b30392ec34a749435aafd3ed21645 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b274be331bfb416a9c691baf91b554dc : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57da2bebee37497b90fa915c2505e4ea : POST_FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a6e0e31ba25abcaca7d77ab4fd4ba88fb1b53e25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 15:52:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"69c177cfa98840cf022e939468a1be06e1b58c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:10:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2bcf275dc7264f12bdcdb3dc18f36029 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aeb5c44b925444e28d50f98a8e2262e0 : FAILURE in 12m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d15b46269ee7489e868345b2beaf7f68 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42d328e96e94c1bac2c0462f6979503 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86cb3bbd02d54e968d64b38577e3cb16 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/242ef99ea146478aafbc0aaa2323196d : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca66055910024ead8fc677a0a08b60f0 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebd81dfaf8bb45cc8c3df215b318570a : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1c5182b255423fbd3a14570153d62dff0c34d8df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-22 15:14:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"847140a582cc86b7c21da7448df37fb504e9f7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:43:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/570938092d5b4191a1bd15030e9acf68 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04ffea93869347149fa22eb0d624d11c : SUCCESS in 1h 19m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec95e6c4ca6849babc504b6d6ff32b33 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085e26c14d47489f85063fd2c950a656 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a527d1233fe449f8d1eb7e7a03ed40a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bf98692db76408bba29dde73af051e1 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e650ffefd6b4506b6d2b646d9348850 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e2f889e371c4870a24e23df41b7f8af : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"be6f24a22b14ac9ba88e84b5f4b21120ebe1ba11","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-23 08:06:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"470f388c1b9c1c3e1137429191a216be5388969e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-28 15:18:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e3091cd30e9fa57e9a375dab921884ec4d6e93ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-05 15:43:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ef664128c510a98da5802b4563214e7d61d099c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 15:43:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2082588b3cd3b3147bb670c3bbea49a68b2cb3ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:59:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"04f528ce54cb820fc1f425837daaa76fc0cd3976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:59:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff371e43c837464d8e2146b4365267e9 : SUCCESS in 3m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54ed93ca34d442759756179f77d6b5dd : SUCCESS in 1h 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/852000e32d834c7489ac9cf41ac90e2c : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda3fec128b94eabb4531da3caa123bf : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f08e2c3c2684454aaf867b47e17a8828 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fff30f8c21fd4759b7cef04dae0fbdbf : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdf09b7d77a42a493025110d334e854 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc1eec6a219b4e12bf7f6f03baff5750 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e4170b547529b73e707b507e7a0ea22105460eb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 17:01:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7540ba86ff644fcdaf16bfa27b3d5677 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ffc194631124861b77dc64fb1cf5809 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de9b2fc6a184611531c46d519797c428a0aeaead","revisions":{"058587d78dc14c9f6edc84af554a81d3b915fdd7":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:40:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/53/742153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/53/742153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/53/742153/1"}}},"commit":{"parents":[{"commit":"e9204ee7b702f67c4de1c777db9482b065126407","subject":"add property methods to extension for more entry point values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e9204ee7b702f67c4de1c777db9482b065126407"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:40:33.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:40:33.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: I045fa1cd7932d960e1cf49b2c335c7acdeba0b46\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/058587d78dc14c9f6edc84af554a81d3b915fdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/058587d78dc14c9f6edc84af554a81d3b915fdd7"}]},"branch":"refs/heads/master"},"de9b2fc6a184611531c46d519797c428a0aeaead":{"kind":"REWORK","_number":2,"created":"2020-09-15 11:41:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/53/742153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/53/742153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/53/742153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/53/742153/2"}}},"commit":{"parents":[{"commit":"e9204ee7b702f67c4de1c777db9482b065126407","subject":"add property methods to extension for more entry point values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e9204ee7b702f67c4de1c777db9482b065126407"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:40:33.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-15 11:40:37.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules.\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker\n  (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings\n  (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I045fa1cd7932d960e1cf49b2c335c7acdeba0b46\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/de9b2fc6a184611531c46d519797c428a0aeaead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/de9b2fc6a184611531c46d519797c428a0aeaead"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
