)]}'
{"id":"openstack%2Fstevedore~747551","triplet_id":"openstack%2Fstevedore~master~Ic2aaf5f467bcecaec6adff1ead8e159ae749ac50","project":"openstack/stevedore","branch":"master","topic":"bug/1892610","hashtags":[],"change_id":"Ic2aaf5f467bcecaec6adff1ead8e159ae749ac50","subject":"Fix the bug 1892610. There\u0027re some syntax errors in the comment of stevedore code.","status":"MERGED","created":"2020-08-23 02:33:25.000000000","updated":"2020-08-24 11:00:16.000000000","submitted":"2020-08-24 10:58:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747551-1598266727173-7a2d7593","meta_rev_id":"0dc110fc2ac0d0d0d57e7a0f03c5d561b3b6751a","_number":747551,"virtual_id_number":747551,"owner":{"_account_id":28873,"name":"zzwfree","email":"zzhiwen@cn.ibm.com","username":"zzwfree-005"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},{"value":0,"date":"2020-08-23 02:33:38.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 10:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"recommended":{"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},"all":[{"value":1,"date":"2020-08-24 01:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-08-23 13:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-08-24 09:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-23 14:01:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2020-08-24 09:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2020-08-23 07:12:30.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-23 02:33:38.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-08-23 07:12:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-08-23 13:36:20.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-08-23 14:01:29.000000000","updated_by":{"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"},"reviewer":{"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"},"state":"REVIEWER"},{"updated":"2020-08-24 01:00:56.000000000","updated_by":{"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},"reviewer":{"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},"state":"REVIEWER"},{"updated":"2020-08-24 09:39:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-24 10:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e7b51809d3931030d0f211af59aa7ca982f0e5c","author":{"_account_id":28873,"name":"zzwfree","email":"zzhiwen@cn.ibm.com","username":"zzwfree-005"},"date":"2020-08-23 02:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2be15d4fc21956e04dba18dcb0326ac9b4b47800","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-08-23 02:33:38.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"b61151deeb29299223bf284688ef4e313b44c1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 03:43:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c566c99dc74403dacefb9b7c089b94f : SUCCESS in 1h 09m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd12280a211c4725a05750f4518f9ae8 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9344b72f876347bab2a317efc02d0a88 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/258ab153306347fd938e29708054ab5f : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbca54aa68434d75b3e6b146c70dfdf7 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3af1f7a87c546b7b0792ed9a524d275 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d0f162e6a39c7a54ecf0904503fc4a3213a91af7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-08-23 13:36:20.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"0866081908622b8cf90f86bbafcc22d79cde97d2","author":{"_account_id":9725,"name":"Dong Ma","email":"winterma.dong@gmail.com","username":"larainema"},"date":"2020-08-23 14:01:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2f34f58478ce165fa1af628a49c582788c3a5ab1","author":{"_account_id":26338,"name":"changzhi","email":"changzhi1990@gmail.com","username":"ZhiChang"},"date":"2020-08-24 01:00:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"24f94e69845b382c91bbac13238deed3daf6f98b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-24 09:39:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f57aec79d2124239188f513b4125d0ef0f4ffbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 09:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cd602e3328bcd34714df2ecd55f964587715b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 10:58:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03bc8a80354448c496ea286c838a50ec : SUCCESS in 1h 14m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31699794cf2c4af0869b30289d63dbf9 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e885777bc54826ae060fe1f8c9accd : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb96fa8003d4472b95e064fbdc5d62a0 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a131d01c497b437bbd83d39ded566af6 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ea3ac16aa04f3787bae076e909fd15 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"73c86ae92276654085a1c8c1c8b937c492ea36d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 10:58:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67d2a787d89bb661b083a35d4eb63be0ff2011a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:00:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c0d53f2e1a4bd4a37a99c41d45f0d1 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb338e00b7c3e9a53ffa0138eefa2440f4f30f7","revisions":{"efb338e00b7c3e9a53ffa0138eefa2440f4f30f7":{"kind":"REWORK","_number":1,"created":"2020-08-23 02:33:25.000000000","uploader":{"_account_id":28873,"name":"zzwfree","email":"zzhiwen@cn.ibm.com","username":"zzwfree-005"},"ref":"refs/changes/51/747551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/51/747551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/747551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/747551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/747551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/51/747551/1"}}},"commit":{"parents":[{"commit":"e9204ee7b702f67c4de1c777db9482b065126407","subject":"add property methods to extension for more entry point values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/e9204ee7b702f67c4de1c777db9482b065126407"}]}],"author":{"name":"zhangziwen","email":"zzhiwen@cn.ibm.com","date":"2020-08-23 02:26:47.000000000","tz":480},"committer":{"name":"zhangziwen","email":"zzhiwen@cn.ibm.com","date":"2020-08-23 02:27:26.000000000","tz":480},"subject":"Fix the bug 1892610. There\u0027re some syntax errors in the comment of stevedore code.","message":"Fix the bug 1892610. There\u0027re some syntax errors in the comment of stevedore code.\n\nThere is a syntax error in the comment of stevedore code.\nFor example, in the HookManager class:\n    :param on_load_failure_callback: Callback function that will be called when\n        a entrypoint can not be loaded.\nHere should not be `a entrypoint`, just `an entrypoint` instead.\n\nFixes: bug #1892610\n\nChange-Id: Ic2aaf5f467bcecaec6adff1ead8e159ae749ac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/efb338e00b7c3e9a53ffa0138eefa2440f4f30f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/efb338e00b7c3e9a53ffa0138eefa2440f4f30f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
