)]}'
{"id":"openstack%2Fstevedore~751044","triplet_id":"openstack%2Fstevedore~master~Ic483ac68027505402ba32d7f612631e15a678d09","project":"openstack/stevedore","branch":"master","hashtags":[],"change_id":"Ic483ac68027505402ba32d7f612631e15a678d09","subject":"Fix cache dir flooding when running from /tmp","status":"MERGED","created":"2020-09-10 16:25:54.000000000","updated":"2020-11-05 22:11:56.000000000","submitted":"2020-11-05 22:10:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751044-1604614211716-4fbb7558","meta_rev_id":"959f5861c73dcabcc27220834053068b9d282baa","_number":751044,"virtual_id_number":751044,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 22:10:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 10:26:42.000000000","_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-11-05 20:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-09-16 10:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-11-05 20:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-09-16 10:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 10:24:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-05 20:34:52.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-11-05 22:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"543c4cd14c125cb81351cdb1152f4713d91475e0","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-10 16:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f81a592de5bdb034110f3c3c977d3495937ea2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dd3c2b464aa4219a7a52d1d03828690 : SUCCESS in 1h 26m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2160022a4614d1799e766fd8c504220 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d8de32e184043119d128c7fedd04686 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1485108f1b04472696cc808e6cadf5c4 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/224d27fe19c54b55ad579eb42b4ce485 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5520f370f1e45f2ad22f332d7a41994 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ba11f07e804e80b1ebd64fc16019a4 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6b727b526e3408b51edf4a41c3c3c7234d633882","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 13:50:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adbb69506eeaeef9c4e2debb3fabc01adff9ab08","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-14 14:06:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57a5eff73008ed0e3d425268f14f9a055e4c4439","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-14 14:07:07.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nyou are right","accounts_in_message":[],"_revision_number":2},{"id":"a6fd513de39757e0d7b0ef58041d2a0e444503a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:30:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3125bbacb144e958a41dfc2bbcc21ae : SUCCESS in 1h 18m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/804885487f0143688e9a9d32cc5cfa8f : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ceddc354d424ad79097d4078735a1a0 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a92d09b019242abbe2ca50acce8f230 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096394f8d9dc4d10b85a9fa78d2608df : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c993b824cff442b3a07ffb92429a74bf : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774ad3170122466591d20f1cba3884b3 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5c52ff5667a9a22e1cfe5dd12947548ce8501662","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 10:24:31.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nI suppose we want to backport these changes, right?","accounts_in_message":[],"_revision_number":2},{"id":"ea34e5992dd22a8ee137606ece47f51e4b46d028","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-16 10:26:42.000000000","message":"Patch Set 2:\n\n\u003e I suppose we want to backport these changes, right?\n\nwould appreciate, since issue is potentially harmful (few times caused disk full until found the cause)","accounts_in_message":[],"_revision_number":2},{"id":"7683f89977f3ab63f925dccb15e110eaafad27a3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-11-05 20:34:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50720121a3250f96b56bdb5682f1d315e19a71a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 20:35:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fd66463953b5bad305d35386d953737ef5572cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:10:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cbf909c02784f06831f17275758d206 : SUCCESS in 1h 28m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a6b170c50d04dc2b8cb5db5ba3a5fe2 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb8582e016d4939a49b900601f4e536 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e5c8a2b00d48b7a1ad724f37813b74 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f06afa432545be8a62f11a20a391ec : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca3af4996b347869bc29b536c69c67b : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f46f9758bf14b57835988991a1f03da : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f3da9c61bdafc3d885258809b39e2f1ad20a475a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:10:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"959f5861c73dcabcc27220834053068b9d282baa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:11:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0db1332c19340d696fa2e7049f46e45 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80882569e3b249d7801ec24d86773174 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a46e291416f453c612fe27a31d0b57918aaf276","revisions":{"d7aa172202b3a97094ca0f2bad0a11e33d036fa0":{"kind":"REWORK","_number":1,"created":"2020-09-10 16:25:54.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/751044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/44/751044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/44/751044/1"}}},"commit":{"parents":[{"commit":"274eaa6c5487124fae3a19fa8aa80ea7d889105b","subject":"fix supported python versions in documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/274eaa6c5487124fae3a19fa8aa80ea7d889105b"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-10 16:21:37.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-10 16:25:39.000000000","tz":120},"subject":"Fix cache dir flooding when running from /tmp","message":"Fix cache dir flooding when running from /tmp\n\nAnsible invokes modules with executable placed under /tmp. This\ncauses stevedore caching to create bazillions of useless cache files.\nWhen we can identify we run with executable under /tmp or if in the\ntarget cache directory an empty file \u0027.disable\u0027 is present - skip\nwriting cache.\n\nChange-Id: Ic483ac68027505402ba32d7f612631e15a678d09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d7aa172202b3a97094ca0f2bad0a11e33d036fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d7aa172202b3a97094ca0f2bad0a11e33d036fa0"}]},"branch":"refs/heads/master"},"3a46e291416f453c612fe27a31d0b57918aaf276":{"kind":"REWORK","_number":2,"created":"2020-09-14 14:06:26.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/751044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/44/751044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/44/751044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/44/751044/2"}}},"commit":{"parents":[{"commit":"274eaa6c5487124fae3a19fa8aa80ea7d889105b","subject":"fix supported python versions in documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/274eaa6c5487124fae3a19fa8aa80ea7d889105b"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-10 16:21:37.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-14 14:06:14.000000000","tz":120},"subject":"Fix cache dir flooding when running from /tmp","message":"Fix cache dir flooding when running from /tmp\n\nAnsible invokes modules with executable placed under /tmp. This\ncauses stevedore caching to create bazillions of useless cache files.\nWhen we can identify we run with executable under /tmp or if in the\ntarget cache directory an empty file \u0027.disable\u0027 is present - skip\nwriting cache.\n\nChange-Id: Ic483ac68027505402ba32d7f612631e15a678d09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3a46e291416f453c612fe27a31d0b57918aaf276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3a46e291416f453c612fe27a31d0b57918aaf276"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
