)]}'
{"id":"openstack%2Fstevedore~761143","triplet_id":"openstack%2Fstevedore~master~Ieb5fd2c1efd9297b9b589e6f13cbe5e0a4cb73cb","project":"openstack/stevedore","branch":"master","topic":"tox-python38","hashtags":[],"change_id":"Ieb5fd2c1efd9297b9b589e6f13cbe5e0a4cb73cb","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2020-11-03 13:32:30.000000000","updated":"2020-11-06 11:44:55.000000000","submitted":"2020-11-06 11:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761143-1604663009160-294064b8","meta_rev_id":"4545236f60c27e0120ad8e16e0ccddaa4979dcec","_number":761143,"virtual_id_number":761143,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-11-06 08:50:34.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 11:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-03 15:26:26.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-03 16:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-03 16:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 15:26:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-03 16:00:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-11-06 11:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c537b337551830d6365764e300877eda8602f979","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-03 13:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f49bfc2c791c76b9f2d52ccce223ae074edfe68f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 13:58:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36c52ccb23ba857edb6f8af13c8e479b1599f6e7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 14:30:19.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nActually, maybe we want to use py3 (suggested by smcginnis) to avoid this tedious work every cycle?","accounts_in_message":[],"_revision_number":1},{"id":"0c19c71a97f9882e11743746df42e90db1465de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 15:08:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e81fb3fc22e84cd28143a4b679c0e22d : SUCCESS in 1h 27m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0704b52970b643299bc71d1737ccc06c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5850fe390305486195dcc727b0f72d9e : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11011afefaa841e1bd5c4b1063a3ac75 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c006103fddc47208e64a67807f66f73 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7527c45d99b43c8ba0adf9de89394f4 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c675a3357554c80a274a90eedd2a18d : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e4f3ca923a4193837e5cfbaa9f66e2 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"47d04266db3112c347dd802016e96eae941c656c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-03 15:14:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2de743bf24d9da2a882fb5daad8378568eb379f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-03 15:26:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6dc6fbd070c4e5189f254bb4553750174ac50edb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 16:00:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d324b62adbb5c6aa28cdac4c8cc996cf77cb301b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 17:00:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81d4a484ae514f4db0f92ca08a31456d : SUCCESS in 1h 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/402a6e8142f94704a8606e4c52a370dc : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530be0e0b5684cb4a08e9df2f5c136d1 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26ef7b76e37246938fb926da805c7de5 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/277968d8d42245a19cf413996a6c3bc9 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8281dd8a8d16464c98b7b65449c2be2c : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4829a3cf66984308b231292218add8b1 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b51c47653a3413094d5072437781568 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f58d2fd2524aa8eb5d3b3a8412fe9ab129b9e681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 17:00:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d3072d6f50c2cf42759a8c71ef32492a174a341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:12:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90429d9cc0fe42cab414568d555cc864 : FAILURE in 1h 05m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af12b11b968d4ed0b07fd4b5ac976f85 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e235cdf7d2b44f30907064351baa6d84 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9aa680f58f7446e87c8e326dab84f4e : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20ab264410754649bf0a1fec200c368b : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84dbaf83150470e857f898bdbe7c0b3 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f73644b5e647a2972c7c8995f3ef50 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"cd5dcc5c1a4c18a3854c3bb0b677ccf1ac78055b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 08:53:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5d730ea90c31d0be4b9c57f8144fe73cd0ad85ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:07:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1758dff41704010aa19fc915b325992 : FAILURE in 1h 08m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c12bad7c6e7e4516b777991679e90e5b : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc42873c6a3141859fcc96f04b1006f1 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8751b70c39ec4b27bb76aa404d6d200b : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f26ec3c699f40eb929cd7c86ac67382 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd902c1312e94ef6b0d843d5f65cdacc : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bab7f666c6e49a5bfb6312dc317b1fb : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/defcb7af2ab14c4b8f2f59756e73bcb3 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"bb0287630d9dae0a1bb2af1f3a57bff85dd253fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 12:10:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"34837e8c879aa58dbffd549557a6bcbbc233074b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:39:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe872afa50ba451ba7a91693778f2f2a : SUCCESS in 1h 27m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51111b00a30944c4bbce4735992e01a5 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4571056bf04eaa91ddc75e8fd023a7 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9380306899d440f6a0febb5f1e04f4e2 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f4dcb628324afc95c3c109143aa2e3 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b6ffef17b94472a987d86b52aa6ae48 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675e5e919d824dd0ac9c3db201215d7f : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05126e2deef74be89777e09f9c7c24bb : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ec46daaf4456a27e3fe0c53529f04f570d17f216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:39:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0eaa32b88680a36ab03d702bb44c1d799fc3a430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 14:01:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59300e0bd98744259addb42064174b38 : FAILURE in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fffc8b2255d4732a8dd74893046d68c : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4cfb20449bf4615ba7f791e048fc84d : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a539444e0748f392157277badac81c : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd1ab1ce255e48cfa3a29f7d77aa8511 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946758a47e9346de813ad09dd599bd4d : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c56e1457874e478eba7a70a6907a770b : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3eb03ad1ebc77326a6918732f68ac1b416c5402a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-06 08:50:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c1947e24eac926a78d94b10ee1e33152049ebbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:12:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fdc2dee33304a1ea7e0ef3be089b89e : SUCCESS in 1h 20m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92dfdf995a704b94b2a2e6eb3f31a870 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4a4ea8b9f249cebc00040632b33e86 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9288804d820f47b98ae2fd5390e635d5 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5fda183d7364fb4a7fadac18ae90e64 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26a64635b095480e88c610e20163f087 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa0bdca4e4d41d6807b108f12161244 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8f3a27b8076446ba188d2cae7c8435f : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"25bf54ddbbcb42396e1736216def6d922cb59ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:13:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1e6dda9d44260038b193a96ee938e6bd02fc849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:43:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b393c5916fcd4caeb9a59af349e8b8da : SUCCESS in 1h 28m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c4a135189dc4215933d493d7f0e92c7 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560b3c51f8374cef8eda79d3b6877a60 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f939d00bb053492d93146c1590516688 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4094e5f5f30142f18c7d332dae1cae54 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289271e9c80a41a8af965965583a3b68 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2908e55efce9445cbf6e0c687096a073 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"07754ebcde89962ff39a86cb01b8a20b448fe83f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:43:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"73c637276baed0b3fba82f3a63eb6ed22c729f2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:44:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1265944c3a4b67a7052e8292002aa9 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f421320d2e4240669c139dddf88ca1d8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41a4d63d388087a12367fb69d449954dbe928576","revisions":{"4b5c55423d75c27be9a0b8f73ca8b024bd13e786":{"kind":"REWORK","_number":1,"created":"2020-11-03 13:32:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/761143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/43/761143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/43/761143/1"}}},"commit":{"parents":[{"commit":"9f4a7d1e26dc426244b5f049ce45b9035bd27c88","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9f4a7d1e26dc426244b5f049ce45b9035bd27c88"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:32:24.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:32:24.000000000","tz":60},"subject":"Use Python 3.8 as the default runtime for tox","message":"Use Python 3.8 as the default runtime for tox\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\nMoving on the most recent officially supported runtime as the default runtime\nfor tox.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Ieb5fd2c1efd9297b9b589e6f13cbe5e0a4cb73cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4b5c55423d75c27be9a0b8f73ca8b024bd13e786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4b5c55423d75c27be9a0b8f73ca8b024bd13e786"}]},"branch":"refs/heads/master"},"41a4d63d388087a12367fb69d449954dbe928576":{"kind":"REWORK","_number":2,"created":"2020-11-03 15:14:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/761143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/43/761143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/43/761143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/43/761143/2"}}},"commit":{"parents":[{"commit":"9f4a7d1e26dc426244b5f049ce45b9035bd27c88","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9f4a7d1e26dc426244b5f049ce45b9035bd27c88"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:32:24.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 15:14:05.000000000","tz":60},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update this at\neach new cycle.\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Ieb5fd2c1efd9297b9b589e6f13cbe5e0a4cb73cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/41a4d63d388087a12367fb69d449954dbe928576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/41a4d63d388087a12367fb69d449954dbe928576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
