)]}'
{"id":"openstack%2Fstevedore~860111","triplet_id":"openstack%2Fstevedore~master~Id962cce270cfbe0517da36babcf123137d803082","project":"openstack/stevedore","branch":"master","topic":"remove-caching","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-23 15:28:32.000000000","reason":"Change was abandoned"},"2472":{"account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"last_update":"2022-10-04 16:56:45.000000000","reason":"\u003cGERRIT_ACCOUNT_2472\u003e replied on the change","reason_account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}}},"hashtags":[],"change_id":"Id962cce270cfbe0517da36babcf123137d803082","subject":"Remove caching","status":"ABANDONED","created":"2022-10-03 16:23:21.000000000","updated":"2023-01-23 15:28:32.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5a6c764d0267a2ff71d7c1a9739e45397f5313d6","_number":860111,"virtual_id_number":860111,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-04 18:35:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2022-10-04 16:27:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-03 16:24:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2022-10-03 18:24:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2170dc256b8e5952d14768bd217ac2e0df861c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-03 16:23:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ef6d827d4ca380a7292d6841d6eb758e53857f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-03 16:27:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"957eddcf05f3e429bc71cfce04519eba6597cd71","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-03 16:48:40.000000000","message":"Topic bug/1966040 removed","accounts_in_message":[],"_revision_number":2},{"id":"856ce49746d02489923ab86fbbd3c98e47d43e70","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-03 16:48:43.000000000","message":"Topic set to remove-caching","accounts_in_message":[],"_revision_number":2},{"id":"e9118fe764cb0339a34a6ace180799559ae8a87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 18:24:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f80b967a2ce24b0295dcbb9df64d0f0b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0325c32579284e98a9b59592aecbabd1 : FAILURE in 2m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66057682306640c0825541c28494bf10 : SUCCESS in 1h 50m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0585a1c1f914cb9a51b46483f2b2e13 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec1772a629ef49d9aeac2f2ac31ee2fc : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e6aab7d0a4242989dd5d541e16c018b : FAILURE in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf5b91aa2c440498df354570d15495f : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cad0ae16914f878dadbb9e3dbc7870 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e215c8ff0b1c4c3a90b586c519e61b0c : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f37ab4569e86a724de86759130d5134e2332d4e9","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-10-03 18:49:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8df31555c1422b4ce5961aeb0663bac4ef9e2ed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 10:14:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8158683b29aaf68375dc970811600969c622c271","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-10-04 14:06:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae57a2c50d7f9a93574d188f3dc20f64311797c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 14:58:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f36ba797b9ba33dc0029b2ef40168ba50b27f80c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 16:26:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6217782abcfc760b0d072676a1526df8f3bf0f78","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 16:27:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8d9e5b4e2d26484f4822a64aded79df0ecb2a20e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 16:29:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d400f0dc0b7d599184fd77b48def9da135066067","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-10-04 16:56:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"463d6720f71e6c96978fb5a025b0a10765387827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 18:35:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/565906eabda24de3904e5d8801281157\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5e6425323ba4f7898317922a40a9fb4 : FAILURE in 2m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03586a43f2ef42539d0cd38e5b40b9d4 : SUCCESS in 2h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26ffcf01ad44ba4a055c4720491539e : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1f35204d014f1d90599cc6ae8d871a : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cafb44eba56d4a87a8a7a73d23baef53 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22ada1f7fd2a42269be74b250b9faa3a : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c70b0b8381944cd902b772ebcd40f65 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d343fa3b554987b5c4462df9ef22a7 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5a6c764d0267a2ff71d7c1a9739e45397f5313d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-23 15:28:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f8d1936860543b2b77ad2131b3e7ce451a5628d1","revisions":{"ef9facd075238cbfb6b54fa09bee63eaab5dbbd0":{"kind":"REWORK","_number":1,"created":"2022-10-03 16:23:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/860111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/11/860111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/11/860111/1"}}},"commit":{"parents":[{"commit":"7a56521dfe9b6c4ca3af381f1db3b50cca72f61e","subject":"cache: Remove unnecessary argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7a56521dfe9b6c4ca3af381f1db3b50cca72f61e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:59:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 16:19:01.000000000","tz":60},"subject":"Remove caching","message":"Remove caching\n\nAs jaraco notes in bug 1942209 [1]:\n\n  In the later releases of importlib_metadata [ref], performance of\n  loading entry points has been dramatically improved, including much\n  faster discovery/loading and an in-memory cache for entry points for a\n  running process.\n\n  These performance gains are measured and tracked in the CI runs for\n  importlib_metadata.\n\n  If possible, this project should consider removing its own cache in\n  favor of the built-in caching in importlib_metadata (also\n  importlib.metadata in Python 3.10). Doing so should also address\n  #1627906 and would limit the dependencies on the data model (such as\n  construction of EntryPoint objects).\n\n  [ref] https://importlib-metadata.readthedocs.io/en/latest/history.html\n\nimportlib-metadata clearly isn\u0027t a stable API so anything that allows us\nto reduce the amount of APIs we are exposed to is a good thing. This\ndoes require reintroducing the need for importlib-metadata (the third\nparty lib) in some of the older versions of Python but that\u0027s a\ntemporary measure at best that will eventually result in us being closer\nto upstream.\n\n[1] https://bugs.launchpad.net/python-stevedore/+bug/1942209\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Id962cce270cfbe0517da36babcf123137d803082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ef9facd075238cbfb6b54fa09bee63eaab5dbbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ef9facd075238cbfb6b54fa09bee63eaab5dbbd0"}]},"branch":"refs/heads/master"},"d86cdc79658a195a4c228b24bb731799e56d03e6":{"kind":"REWORK","_number":2,"created":"2022-10-03 16:27:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/860111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/11/860111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/11/860111/2"}}},"commit":{"parents":[{"commit":"1d2055fe96a13dff4a4711dc13ab5792ccf09541","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1d2055fe96a13dff4a4711dc13ab5792ccf09541"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:59:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 16:26:56.000000000","tz":60},"subject":"Remove caching","message":"Remove caching\n\nAs jaraco notes in bug 1942209 [1]:\n\n  In the later releases of importlib_metadata [ref], performance of\n  loading entry points has been dramatically improved, including much\n  faster discovery/loading and an in-memory cache for entry points for a\n  running process.\n\n  These performance gains are measured and tracked in the CI runs for\n  importlib_metadata.\n\n  If possible, this project should consider removing its own cache in\n  favor of the built-in caching in importlib_metadata (also\n  importlib.metadata in Python 3.10). Doing so should also address\n  #1627906 and would limit the dependencies on the data model (such as\n  construction of EntryPoint objects).\n\n  [ref] https://importlib-metadata.readthedocs.io/en/latest/history.html\n\nimportlib-metadata clearly isn\u0027t a stable API so anything that allows us\nto reduce the amount of APIs we are exposed to is a good thing. This\ndoes require reintroducing the need for importlib-metadata (the third\nparty lib) in some of the older versions of Python but that\u0027s a\ntemporary measure at best that will eventually result in us being closer\nto upstream.\n\n[1] https://bugs.launchpad.net/python-stevedore/+bug/1942209\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Id962cce270cfbe0517da36babcf123137d803082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d86cdc79658a195a4c228b24bb731799e56d03e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/d86cdc79658a195a4c228b24bb731799e56d03e6"}]},"branch":"refs/heads/master"},"13242b8e133e091856794b5eb72f7f5c2adea34d":{"kind":"REWORK","_number":3,"created":"2022-10-04 14:58:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/860111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/11/860111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/11/860111/3"}}},"commit":{"parents":[{"commit":"28fc7164dace83abc3ecc06fcb56f7ca880d735a","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/28fc7164dace83abc3ecc06fcb56f7ca880d735a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:59:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 14:51:22.000000000","tz":60},"subject":"Remove caching","message":"Remove caching\n\nAs jaraco notes in bug 1942209 [1]:\n\n  In the later releases of importlib_metadata [ref], performance of\n  loading entry points has been dramatically improved, including much\n  faster discovery/loading and an in-memory cache for entry points for a\n  running process.\n\n  These performance gains are measured and tracked in the CI runs for\n  importlib_metadata.\n\n  If possible, this project should consider removing its own cache in\n  favor of the built-in caching in importlib_metadata (also\n  importlib.metadata in Python 3.10). Doing so should also address\n  #1627906 and would limit the dependencies on the data model (such as\n  construction of EntryPoint objects).\n\n  [ref] https://importlib-metadata.readthedocs.io/en/latest/history.html\n\nimportlib-metadata clearly isn\u0027t a stable API so anything that allows us\nto reduce the amount of APIs we are exposed to is a good thing. This\ndoes require reintroducing the need for importlib-metadata (the third\nparty lib) in some of the older versions of Python but that\u0027s a\ntemporary measure at best that will eventually result in us being closer\nto upstream.\n\n[1] https://bugs.launchpad.net/python-stevedore/+bug/1942209\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Id962cce270cfbe0517da36babcf123137d803082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/13242b8e133e091856794b5eb72f7f5c2adea34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/13242b8e133e091856794b5eb72f7f5c2adea34d"}]},"branch":"refs/heads/master"},"f8d1936860543b2b77ad2131b3e7ce451a5628d1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-04 16:26:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/860111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/11/860111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/11/860111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/11/860111/4"}}},"commit":{"parents":[{"commit":"7eca3f5661bb7624139d629c097167c82ce4692c","subject":"Remove Extension.extras","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7eca3f5661bb7624139d629c097167c82ce4692c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:59:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 16:26:32.000000000","tz":60},"subject":"Remove caching","message":"Remove caching\n\nAs jaraco notes in bug 1942209 [1]:\n\n  In the later releases of importlib_metadata [ref], performance of\n  loading entry points has been dramatically improved, including much\n  faster discovery/loading and an in-memory cache for entry points for a\n  running process.\n\n  These performance gains are measured and tracked in the CI runs for\n  importlib_metadata.\n\n  If possible, this project should consider removing its own cache in\n  favor of the built-in caching in importlib_metadata (also\n  importlib.metadata in Python 3.10). Doing so should also address\n  #1627906 and would limit the dependencies on the data model (such as\n  construction of EntryPoint objects).\n\n  [ref] https://importlib-metadata.readthedocs.io/en/latest/history.html\n\nimportlib-metadata clearly isn\u0027t a stable API so anything that allows us\nto reduce the amount of APIs we are exposed to is a good thing. This\ndoes require reintroducing the need for importlib-metadata (the third\nparty lib) in some of the older versions of Python but that\u0027s a\ntemporary measure at best that will eventually result in us being closer\nto upstream.\n\n[1] https://bugs.launchpad.net/python-stevedore/+bug/1942209\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Id962cce270cfbe0517da36babcf123137d803082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/f8d1936860543b2b77ad2131b3e7ce451a5628d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/f8d1936860543b2b77ad2131b3e7ce451a5628d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
