)]}'
{"id":"openstack%2Fstevedore~860269","triplet_id":"openstack%2Fstevedore~master~Id772d9f002e6945666685138bdef8f8ca32b5229","project":"openstack/stevedore","branch":"master","topic":"remove-caching","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-23 16:01:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id772d9f002e6945666685138bdef8f8ca32b5229","subject":"Remove Extension.extras","status":"MERGED","created":"2022-10-04 14:58:50.000000000","updated":"2023-01-23 16:02:52.000000000","submitted":"2023-01-23 16:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860269-remove-caching","meta_rev_id":"1595c42a9d3ed514d2dbfc1a5c124a25f08bcf52","_number":860269,"virtual_id_number":860269,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 16:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-23 14:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-10-05 11:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-01-23 14:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 16:58:43.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2022-10-04 18:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 11:40:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-23 14:17:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"a242a55a0ddc2a70610940ef957cb4b77bd9b0d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 14:58:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"626cc67035c3d79a9dc79b3dc2aaf66d550e678d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 16:26:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2d33e4ffea33cbe87736df730852fdcbb1e6b4aa","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-10-04 16:58:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2cbe76023063bfb6516cc8dc7d61a21154fc59d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 18:13:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535d647df7a34142adda426a16084547\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9adc0b397c8047da8ccb92381b3b5cdf : SUCCESS in 1h 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f0d169ac764075b0e52ec34e0228e0 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae9e0f15190445a086a50f85b26232a5 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f57b98a3d5a34166be89a4d6f05d657e : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/939b4731d79f49eb9cbb4edd445c6c35 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b111edb2334de8b4c37a0aa7f85a94 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1f68398aca047d5bf6354ac8e0079cc : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e4204ac5f0aff6c72f1f44bf49b489d341aad085","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-05 11:40:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21fd2944776683eb59e6e9107130893bc36e553b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 14:17:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29bbda90b250c59096915b1cacd39aa6127ed34f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 14:17:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"564c61b377204eb53079610a9a36bad89caeee93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:01:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e646ae8a014fe4a7d109dc9bf9781e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3691988a20d4912a56b1a4c7e342054 : SUCCESS in 1h 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2652aee10c824e5db0dff0fd57aae2da : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668d00942c7948ad896fce8daabdac8a : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1c7f4f25df244ff95cc44cc1b7b54fb : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4385b24f0d446c940058b6c1ec4139 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92cc7f3df3b440b8a5b4cca847714971 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e1e516ab0e7bfb9327db6cd1d4f9a220ad06a5d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1595c42a9d3ed514d2dbfc1a5c124a25f08bcf52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:02:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128db0bb33d4429a9ce60383adddf1d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261c7f5e5fab4847a282b864b829311b : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8fa2dcdbcf640d29cdb0f11bef55c5a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7eca3f5661bb7624139d629c097167c82ce4692c","revisions":{"3228d81faa5a6415523354030ace6409a6bc6b13":{"kind":"REWORK","_number":1,"created":"2022-10-04 14:58:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/860269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/69/860269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/69/860269/1"}}},"commit":{"parents":[{"commit":"eaacd7015a19be1a4c13979be975bc6af6c2ddf6","subject":"fixup! Remove caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/eaacd7015a19be1a4c13979be975bc6af6c2ddf6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 14:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 14:58:42.000000000","tz":60},"subject":"Remove Extension.extras","message":"Remove Extension.extras\n\nThe behavior of the \u0027EntryPoint.extras` attribute that this exposes has\nchanged in recent versions of \u0027importlib_metadata\u0027/\u0027importlib.metadata\u0027.\nIn change Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b we applied a\ntemporary fix to preserve the legacy behavior on these newer versions,\nhowever, given that this is actually a deprecated attribute [1], the\nbest long-term option seems to be not exposing things. Do just that.\n\n[1] https://packaging.python.org/en/latest/specifications/entry-points/#data-model\n\nChange-Id: Id772d9f002e6945666685138bdef8f8ca32b5229\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3228d81faa5a6415523354030ace6409a6bc6b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3228d81faa5a6415523354030ace6409a6bc6b13"}]},"branch":"refs/heads/master"},"7eca3f5661bb7624139d629c097167c82ce4692c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-04 16:26:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/860269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/69/860269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/69/860269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/69/860269/2"}}},"commit":{"parents":[{"commit":"28fc7164dace83abc3ecc06fcb56f7ca880d735a","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/28fc7164dace83abc3ecc06fcb56f7ca880d735a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 14:53:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-04 16:26:32.000000000","tz":60},"subject":"Remove Extension.extras","message":"Remove Extension.extras\n\nThe behavior of the \u0027EntryPoint.extras` attribute that this exposes has\nchanged in recent versions of \u0027importlib_metadata\u0027/\u0027importlib.metadata\u0027.\nIn change Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b we applied a\ntemporary fix to preserve the legacy behavior on these newer versions,\nhowever, given that this is actually a deprecated attribute [1], the\nbest long-term option seems to be not exposing things. Do just that.\n\n[1] https://packaging.python.org/en/latest/specifications/entry-points/#data-model\n\nChange-Id: Id772d9f002e6945666685138bdef8f8ca32b5229\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7eca3f5661bb7624139d629c097167c82ce4692c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7eca3f5661bb7624139d629c097167c82ce4692c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
