)]}'
{"id":"openstack%2Fstevedore~860394","triplet_id":"openstack%2Fstevedore~stable%2Fyoga~Ib9c2b0a14edea91e97d122d2ac93b650029f918e","project":"openstack/stevedore","branch":"stable/yoga","topic":"bug/1966040","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-05 17:49:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9c2b0a14edea91e97d122d2ac93b650029f918e","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","status":"MERGED","created":"2022-10-05 10:16:25.000000000","updated":"2022-10-05 17:50:12.000000000","submitted":"2022-10-05 17:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860394-bug/1966040","meta_rev_id":"facef6aa11ce4e3536087ac2a0046867506d4c08","_number":860394,"virtual_id_number":860394,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 17:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-05 12:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-05 15:45:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 11:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 12:39:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"c625610d36c93d7cc5f2b310869c445ef4cfa454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-05 10:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e71d64040325b7462e893b110e5131127b268238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/108dd8242dbc4545b7bec8c04bfefc8a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd11c43ecce945d79d5e7480cbb60f4f : SUCCESS in 1h 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b0d31ecc6f4c32a3dfaaa7f4ab999b : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d7828e9020a4b619dadfced3bd28c83 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9897ce0240584aa0b0f7f614f7779a2f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65397eb1a0f74f8985e4e830d3ab1788 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a3ed57ac987d79917fb842982e78a529389d144a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-05 12:39:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6672bc345c033b26632b822acacefa5c482bbab8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-05 15:45:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ca6c4e99a226a69ea0de5ec6485dfeb4a4e1915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 15:45:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"818e42867c00ac84fc541bc5e74d13e15f8271be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:49:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80648ce874c14b839ea159f55400cb67\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c9b3071df24b129894424b2db1a71a : SUCCESS in 1h 58m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8443bf76b34e45b698aca06672688a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f269726ce846c299728449091ec02a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc84c9864983473387d12dc54f384077 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2bcc698f8e74904bc7c504d2e8a08c6 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fa8bfeaecd83b8a7b94e458bd4928a1cafbd0b2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:49:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"facef6aa11ce4e3536087ac2a0046867506d4c08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:50:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e110b9ce2444876aafceb52f1af1823\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdeab57dc4854033ad3a7ba6e1b6bffc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"143a3e9f0716690be7343d4d083f65d7624b3d2e","revisions":{"143a3e9f0716690be7343d4d083f65d7624b3d2e":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:16:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/860394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/94/860394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/94/860394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/94/860394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/94/860394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/94/860394/1"}}},"commit":{"parents":[{"commit":"9c78134d7b2a3a355bb2ba155fe80382d375e3ab","subject":"Fix compatibility with Python 3.10, 3.9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9c78134d7b2a3a355bb2ba155fe80382d375e3ab"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:14:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-05 10:16:18.000000000","tz":60},"subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","message":"Fix compatibility with Python 3.12, importlib-metadata 5.0\n\nimportlib-metadata 5.0 has removed support for dict-style interaction\nwith entrypoints [1]. This is going to eventually affect us when Python\n3.12 is released but even before then anyone not properly using upper\nconstraints with an older Python 3.7-based release (the only Python\nversion where we require the third-party importlib-metadata package\nrather than the stdlib importlib.metadata package) will be bitten. Fix\nit now to address both.\n\n[1] https://github.com/python/importlib_metadata/commit/dde2b9de2973ce1c6fa9ba21dfe81069b0baa77b\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9c2b0a14edea91e97d122d2ac93b650029f918e\nCloses-Bug: #1991559\n(cherry picked from commit 28fc7164dace83abc3ecc06fcb56f7ca880d735a)\n(cherry picked from commit 1e4d71d4b552494ff33ad03b76d2a590d58efe77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/143a3e9f0716690be7343d4d083f65d7624b3d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/143a3e9f0716690be7343d4d083f65d7624b3d2e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
