)]}'
{"id":"openstack%2Fstevedore~860395","triplet_id":"openstack%2Fstevedore~stable%2Fxena~Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b","project":"openstack/stevedore","branch":"stable/xena","topic":"bug/1966040","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-18 10:03:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b","subject":"Fix compatibility with Python 3.10, 3.9.11","status":"MERGED","created":"2022-10-05 10:17:13.000000000","updated":"2022-10-18 10:03:58.000000000","submitted":"2022-10-18 10:03:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860395-bug/1966040","meta_rev_id":"4e04700761b9ce6cd0e4533fc5f1525fba6d1ecd","_number":860395,"virtual_id_number":860395,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 10:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-18 08:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-18 08:31:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 11:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 08:31:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"970b2c9e4aafc79b7aa69d34074e30407acefa08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-05 10:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2970a70c2da53d385ce132cd73daf446b46372f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:53:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638287c1d714490eb06bc33055f2f7e7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d518e14db744e4ca1a3148bd4af4fee : SUCCESS in 1h 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356528791fd64e98a51d5e4adf364397 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f63244b1f4d4975963faa0cdef6fa4d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb95eef49ee8473db4a4736e7437eefd : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea4817535d7a4d3d85db76864857a97d : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5201b26babf44240b7ba1ea02657961c : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"08a0a4927277dd3e904941cbfe60a94542d32e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-13 13:42:57.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0cf992949b5cd1a655b46ede1bf634a8a5c48ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 15:19:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab211c274de4f4aa1338938bc8b9362\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd1c1a1cd5f1427798ac15c6f821813a : SUCCESS in 1h 31m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7250866f32404ca28f40de439b5e4380 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6400335d76774ef4b5b516cdf2cd9e26 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575e2c769b4a4def87651dc2c72eb443 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bec05128d284b4bab3d58f0fec0a195 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcdaba16ee724c3295b02afc3e5a47fd : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1af091bed46cada9b9efcf9594464ac350c67741","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-18 08:31:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1bd4277b06bc9e312101e71eea81d5ec71d09cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 08:32:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77a2c991f27c58ddf5d67948c8099fcfe18dd931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:03:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce8a70782b84167a6e60ab1d11427ec\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae5379c058be4e0aa22bc8fca4a6e210 : SUCCESS in 1h 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca572e34a90d4f0e82889d5d7859751a : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2c73520be98412f920461eeb27071ff : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/448a9bfa7dfd4db18cd7ebd6bb10355c : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c40b6090584d7db31eb4b5c07a2927 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"1ecd36485abc95690e185d52d830852258a8f6da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:03:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4e04700761b9ce6cd0e4533fc5f1525fba6d1ecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:03:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5212be77d1f04c9a9870e5f6ccc4bdf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d31183c41544e57b219f12868831e17 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52433382eed2e4f1783fe3a14813d2409b122d41","revisions":{"b006bcd083929145fe48ff3fdb3ca0732763924e":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:17:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/860395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/860395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/860395/1"}}},"commit":{"parents":[{"commit":"cb128a8abed1e05b3ff06d849dad1dcb499e75cb","subject":"Rely on member access, the preferred access since importlib_metadata 4.8.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/cb128a8abed1e05b3ff06d849dad1dcb499e75cb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-09 15:50:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-05 10:16:57.000000000","tz":60},"subject":"Fix compatibility with Python 3.10, 3.9.11","message":"Fix compatibility with Python 3.10, 3.9.11\n\nA fix to \u0027importlib.metadata\u0027 in Python 3.10 [1], later backported to\n3.9 and released in 3.9.11 [2], has broken our tests. Fix them.\n\n[1] https://github.com/python/cpython/commit/b1e286860742e7ba6fadc75e3ddb6c2899a56919\n[2] https://github.com/python/cpython/commit/177be52517da9a876a3f9e670f88c4731b906986\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #1966040\nChange-Id: Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b\n(cherry picked from commit 663d56c9535c64f4f8bd363abf69406975585ee9)\n(cherry picked from commit 2312415517cd5ded007127d21f75f78ff2f63f52)\n(cherry picked from commit 9c78134d7b2a3a355bb2ba155fe80382d375e3ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/b006bcd083929145fe48ff3fdb3ca0732763924e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/b006bcd083929145fe48ff3fdb3ca0732763924e"}]},"branch":"refs/heads/stable/xena"},"52433382eed2e4f1783fe3a14813d2409b122d41":{"kind":"NO_CHANGE","_number":2,"created":"2022-10-13 13:42:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/860395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/860395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/860395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/860395/2"}}},"commit":{"parents":[{"commit":"cb128a8abed1e05b3ff06d849dad1dcb499e75cb","subject":"Rely on member access, the preferred access since importlib_metadata 4.8.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/cb128a8abed1e05b3ff06d849dad1dcb499e75cb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-09 15:50:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-13 13:42:46.000000000","tz":60},"subject":"Fix compatibility with Python 3.10, 3.9.11","message":"Fix compatibility with Python 3.10, 3.9.11\n\nA fix to \u0027importlib.metadata\u0027 in Python 3.10 [1], later backported to\n3.9 and released in 3.9.11 [2], has broken our tests. Fix them.\n\n[1] https://github.com/python/cpython/commit/b1e286860742e7ba6fadc75e3ddb6c2899a56919\n[2] https://github.com/python/cpython/commit/177be52517da9a876a3f9e670f88c4731b906986\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #1966040\nChange-Id: Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b\n(cherry picked from commit 663d56c9535c64f4f8bd363abf69406975585ee9)\n(cherry picked from commit 2312415517cd5ded007127d21f75f78ff2f63f52)\n(cherry picked from commit 9c78134d7b2a3a355bb2ba155fe80382d375e3ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/52433382eed2e4f1783fe3a14813d2409b122d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/52433382eed2e4f1783fe3a14813d2409b122d41"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
