)]}'
{"id":"openstack%2Fstevedore~860396","triplet_id":"openstack%2Fstevedore~stable%2Fxena~Ib9c2b0a14edea91e97d122d2ac93b650029f918e","project":"openstack/stevedore","branch":"stable/xena","topic":"bug/1966040","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-18 10:07:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9c2b0a14edea91e97d122d2ac93b650029f918e","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","status":"MERGED","created":"2022-10-05 10:17:13.000000000","updated":"2022-10-18 10:08:47.000000000","submitted":"2022-10-18 10:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860396-bug/1966040","meta_rev_id":"5615acd217be2c8a9fcb50f9f78af4de30844f92","_number":860396,"virtual_id_number":860396,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 10:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-18 08:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-18 08:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 11:31:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 08:31:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"35bc9d28976781c4539ff7c65ab496553ce753fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-05 10:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"059e4228b6725294ff0f6862e68515a0e650af05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:31:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/495ec41f788f4bb9be8661f3d5d148c8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16c3c9e24550423b9f9347aa48169997 : SUCCESS in 1h 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd20ea844851481987afcc2f44ba9576 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eaf1ded6149441b9761f3c63dbcf609 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64cbfa91bbd84e75a4add8677c37bcb1 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/883d56d6d570468285f40c04980cf67b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3114128835f468ea09b3a49a70a66e2 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b2e85398fa6e16b9c81203c2276b97aa5c4e5e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-13 13:42:57.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b17bd24634b44bc26a108a418de1f247a87bd386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 15:31:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e17d4f810cc410f86550c1436a45e23\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7795e5d328ed45a1b1eed7090f3ddad8 : SUCCESS in 1h 42m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186542c471b24965bda19e2de3a264ad : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9edf9cf96c2c462bbf59078cb2028511 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ad5f37aeac409baf8f8881d9c4bb0f : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9aa83d59f3b4bbd819817bdfe237f02 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7309a7e062334dc7873085b7be541467 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f08a42273c706e2be604cd133ee6e1f9815303af","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-18 08:31:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"60832c8a2cdef0e736ae96b3988b2c7406b96796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 08:32:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b80fbaa104a39ab3faab6478c713a9a9e426d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:07:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b8b43f552b413ba5f80a3379e084d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bfc376a66044796a6162627c30e641d : SUCCESS in 1h 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75249bdfff494bfe972e5908a3c6a9bd : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13b754d7fc78419bafa125d283b12eb7 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/928db3b9b4a249e48bb5b6e33ee484ab : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4cb4d94fb143fd8810dafb9b2e79a0 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"dad425a3f1fd008b9cc4d4b65cdf7c356fe818f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:07:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5615acd217be2c8a9fcb50f9f78af4de30844f92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:08:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e70fbe4c9b294749a6c7a3a6bd5ef6fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80264a9025dc476faee6eda0f9f08f47 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97dee6c435e4609231c96106fe253a1d9749d9cc","revisions":{"54acda936d41e9de7601b36223050faaccef8529":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:17:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/860396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/96/860396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/96/860396/1"}}},"commit":{"parents":[{"commit":"b006bcd083929145fe48ff3fdb3ca0732763924e","subject":"Fix compatibility with Python 3.10, 3.9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/b006bcd083929145fe48ff3fdb3ca0732763924e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:14:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-05 10:17:08.000000000","tz":60},"subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","message":"Fix compatibility with Python 3.12, importlib-metadata 5.0\n\nimportlib-metadata 5.0 has removed support for dict-style interaction\nwith entrypoints [1]. This is going to eventually affect us when Python\n3.12 is released but even before then anyone not properly using upper\nconstraints with an older Python 3.7-based release (the only Python\nversion where we require the third-party importlib-metadata package\nrather than the stdlib importlib.metadata package) will be bitten. Fix\nit now to address both.\n\n[1] https://github.com/python/importlib_metadata/commit/dde2b9de2973ce1c6fa9ba21dfe81069b0baa77b\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9c2b0a14edea91e97d122d2ac93b650029f918e\nCloses-Bug: #1991559\n(cherry picked from commit 28fc7164dace83abc3ecc06fcb56f7ca880d735a)\n(cherry picked from commit 1e4d71d4b552494ff33ad03b76d2a590d58efe77)\n(cherry picked from commit 143a3e9f0716690be7343d4d083f65d7624b3d2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/54acda936d41e9de7601b36223050faaccef8529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/54acda936d41e9de7601b36223050faaccef8529"}]},"branch":"refs/heads/stable/xena"},"97dee6c435e4609231c96106fe253a1d9749d9cc":{"kind":"NO_CHANGE","_number":2,"created":"2022-10-13 13:42:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/860396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/96/860396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/96/860396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/96/860396/2"}}},"commit":{"parents":[{"commit":"52433382eed2e4f1783fe3a14813d2409b122d41","subject":"Fix compatibility with Python 3.10, 3.9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/52433382eed2e4f1783fe3a14813d2409b122d41"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:14:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-13 13:42:50.000000000","tz":60},"subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","message":"Fix compatibility with Python 3.12, importlib-metadata 5.0\n\nimportlib-metadata 5.0 has removed support for dict-style interaction\nwith entrypoints [1]. This is going to eventually affect us when Python\n3.12 is released but even before then anyone not properly using upper\nconstraints with an older Python 3.7-based release (the only Python\nversion where we require the third-party importlib-metadata package\nrather than the stdlib importlib.metadata package) will be bitten. Fix\nit now to address both.\n\n[1] https://github.com/python/importlib_metadata/commit/dde2b9de2973ce1c6fa9ba21dfe81069b0baa77b\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9c2b0a14edea91e97d122d2ac93b650029f918e\nCloses-Bug: #1991559\n(cherry picked from commit 28fc7164dace83abc3ecc06fcb56f7ca880d735a)\n(cherry picked from commit 1e4d71d4b552494ff33ad03b76d2a590d58efe77)\n(cherry picked from commit 143a3e9f0716690be7343d4d083f65d7624b3d2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/97dee6c435e4609231c96106fe253a1d9749d9cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/97dee6c435e4609231c96106fe253a1d9749d9cc"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
