)]}'
{"id":"openstack%2Fstevedore~860397","triplet_id":"openstack%2Fstevedore~stable%2Fwallaby~Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b","project":"openstack/stevedore","branch":"stable/wallaby","topic":"bug/1966040","hashtags":[],"change_id":"Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b","subject":"Fix compatibility with Python 3.10, 3.9.11","status":"MERGED","created":"2022-10-05 10:18:14.000000000","updated":"2022-10-12 15:38:14.000000000","submitted":"2022-10-12 15:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860397-bug/1966040","meta_rev_id":"2f47d516164b751d528b7bff58b5176e837e782e","_number":860397,"virtual_id_number":860397,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 15:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-10-12 13:56:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-10-12 13:56:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 11:57:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"124b289838820e1974a4ec4ca82f895a369e54c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-05 10:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fedc205c0bd703c66ddea1d19fa1d0646e5c8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8181d0d238442083d470f1b8d21a13\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6c325b395ce4fad82f230c77a56868d : SUCCESS in 1h 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae0ec1ee942417dae2c7d86f634918c : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc1860367c5420f94cab49b61890aa0 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccc388673bba478b9012699d0091b00a : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deac58f979014cafad750e1236b522d9 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cac6ce0ec664e41b400a4a341ef070f : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"02a13118a40fbf9dec84cd064bfd187f1e05b47a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-12 13:56:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"846d0f3828720b968ad757d7698d8ff94f2cb383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 13:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bd285bac48d06a108c4e5aa01fa4f170a0c5434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:37:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5d1cba2e014b059d5c67adc04f8566\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cc9f39127744d82985a76b0ce25e170 : SUCCESS in 1h 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d532cb30799d4d7e8aef553e6c8f9ae6 : SUCCESS in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59105192a6bb41568c41eb91d50120e7 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b383749bb9a646f7a78c03e3ee94c940 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9cb763bd77f4ca1992de27061083a7d : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"627e85199fcca819552680a4704d429792a9691c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:37:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2f47d516164b751d528b7bff58b5176e837e782e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:38:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77737fffc0444bdbd6d466715f2b06a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d92a0455e434b57bd8313c4a1b6979b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab0135e09af9b36403532e86766ac0c89e8016ea","revisions":{"ab0135e09af9b36403532e86766ac0c89e8016ea":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:18:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/860397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/97/860397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/860397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/860397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/860397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/97/860397/1"}}},"commit":{"parents":[{"commit":"eb7b6a70c5f9e34f6640d73fd07e79f33f64a194","subject":"Rely on member access, the preferred access since importlib_metadata 4.8.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/eb7b6a70c5f9e34f6640d73fd07e79f33f64a194"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-09 15:50:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-05 10:17:53.000000000","tz":60},"subject":"Fix compatibility with Python 3.10, 3.9.11","message":"Fix compatibility with Python 3.10, 3.9.11\n\nA fix to \u0027importlib.metadata\u0027 in Python 3.10 [1], later backported to\n3.9 and released in 3.9.11 [2], has broken our tests. Fix them.\n\n[1] https://github.com/python/cpython/commit/b1e286860742e7ba6fadc75e3ddb6c2899a56919\n[2] https://github.com/python/cpython/commit/177be52517da9a876a3f9e670f88c4731b906986\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #1966040\nChange-Id: Iff536d4f4267efbebc4be1e7e5da8a9fde39f79b\n(cherry picked from commit 663d56c9535c64f4f8bd363abf69406975585ee9)\n(cherry picked from commit 2312415517cd5ded007127d21f75f78ff2f63f52)\n(cherry picked from commit 9c78134d7b2a3a355bb2ba155fe80382d375e3ab)\n(cherry picked from commit b006bcd083929145fe48ff3fdb3ca0732763924e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ab0135e09af9b36403532e86766ac0c89e8016ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ab0135e09af9b36403532e86766ac0c89e8016ea"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
