)]}'
{"id":"openstack%2Fstevedore~860398","triplet_id":"openstack%2Fstevedore~stable%2Fwallaby~Ib9c2b0a14edea91e97d122d2ac93b650029f918e","project":"openstack/stevedore","branch":"stable/wallaby","topic":"bug/1966040","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-12 15:37:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9c2b0a14edea91e97d122d2ac93b650029f918e","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","status":"MERGED","created":"2022-10-05 10:18:14.000000000","updated":"2022-10-12 15:38:59.000000000","submitted":"2022-10-12 15:37:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860398-bug/1966040","meta_rev_id":"7cbaf11d37158016c474cf84d3121228e208471f","_number":860398,"virtual_id_number":860398,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 15:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-10-12 13:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-10-12 14:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-10-12 13:56:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 12:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 14:50:37.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"73d16f50a54e6b1c95f749c181b014539f66f20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-05 10:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48c4c9b765e762b4e71b03c20caf02f25491b26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:00:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f426c79365a842968c9293ec9330c2fd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f400d40f34b48b495858692bc855ebe : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd250210b30b4867b2e3d14d9e6a98df : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/342a3d52846142baa790cd8483635589 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec0622e86b344193ac51522cb0e129d4 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7121d6f2583246259af234cddf5201bc : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722800d978f44b138f1d555f4bca23e2 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"28bbe1823edd44ef2b46dc250107dd79cd7fe935","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-12 13:56:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2eb5d9481c4a3e5ac554bf0ae71e6d4179b0f8da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 13:56:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c80ada52615b784e049188d64b4100cc390e91ac","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-12 14:50:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"469e45c630cc8623557dd38b233e01c8ece00fb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:37:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de421c2d28ef4b28aa060f91c1e7f7b2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2998e428c32b47c0b5e3efef9bb2ed54 : SUCCESS in 1h 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c04dc83101b4cd58fca457a28049876 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1431685d2ee4bdbb643b97695bf8cbf : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39f85c2111cf4262b49bff607da5e4fd : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002260ccee3c4d448f0f80db5388a3aa : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0c6a06f24d17ba44bc4bd2f19915b171bf7a787b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:37:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cbaf11d37158016c474cf84d3121228e208471f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 15:38:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/090fc9a0c5ab4679adf6690cae8d2e58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3e9edd07784f1e871d2fd3f5eff065 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340","revisions":{"48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340":{"kind":"REWORK","_number":1,"created":"2022-10-05 10:18:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/860398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/98/860398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/860398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/860398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/860398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/98/860398/1"}}},"commit":{"parents":[{"commit":"ab0135e09af9b36403532e86766ac0c89e8016ea","subject":"Fix compatibility with Python 3.10, 3.9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ab0135e09af9b36403532e86766ac0c89e8016ea"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-03 15:14:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-05 10:18:02.000000000","tz":60},"subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","message":"Fix compatibility with Python 3.12, importlib-metadata 5.0\n\nimportlib-metadata 5.0 has removed support for dict-style interaction\nwith entrypoints [1]. This is going to eventually affect us when Python\n3.12 is released but even before then anyone not properly using upper\nconstraints with an older Python 3.7-based release (the only Python\nversion where we require the third-party importlib-metadata package\nrather than the stdlib importlib.metadata package) will be bitten. Fix\nit now to address both.\n\n[1] https://github.com/python/importlib_metadata/commit/dde2b9de2973ce1c6fa9ba21dfe81069b0baa77b\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9c2b0a14edea91e97d122d2ac93b650029f918e\nCloses-Bug: #1991559\n(cherry picked from commit 28fc7164dace83abc3ecc06fcb56f7ca880d735a)\n(cherry picked from commit 1e4d71d4b552494ff33ad03b76d2a590d58efe77)\n(cherry picked from commit 143a3e9f0716690be7343d4d083f65d7624b3d2e)\n(cherry picked from commit 54acda936d41e9de7601b36223050faaccef8529)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
