)]}'
{"id":"openstack%2Fstevedore~861695","triplet_id":"openstack%2Fstevedore~master~I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","project":"openstack/stevedore","branch":"master","topic":"entrypoints-py37","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-10-19 13:24:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","subject":"Order old importlib-metadata results by group","status":"MERGED","created":"2022-10-18 03:54:28.000000000","updated":"2022-10-19 13:25:48.000000000","submitted":"2022-10-19 13:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"861695-entrypoints-py37","meta_rev_id":"221f4248703f49bea37c3bde41e21e89d3b2635a","_number":861695,"virtual_id_number":861695,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 13:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-10-19 10:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-10-19 12:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2022-10-19 10:16:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-10-19 12:11:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 05:29:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 12:01:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-10-19 10:16:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-10-19 12:11:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"234e80371a02178deffe8098416094d91f6cc6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-10-18 03:54:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e9aeb4ac99e9726bd955c123c780dd9d13aba11","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-10-18 03:56:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bf31cc72695d362a3bdcaf846e07ee285f9bceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 05:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a744908428f94e178e36322003d5f8cb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8c91b3b83994f8285139605b0248e05 : SUCCESS in 1h 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c275115863b4124bc2efd074f1fe77c : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb14683dbb8e4fc5be8c8a60bfa68185 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85946ed50e5748599fe43e33410c8b44 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce02883a1564e4882307c37f0a7959d : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"83d022028070150dd7fef4f04b7958981df5395c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-18 12:01:15.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3200a2a0ed501376421858ed123a25d25e65e23e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-10-18 20:58:42.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fcf0474d9a03a30cf7ac2a2bd5d9125adc9f27ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-10-18 21:00:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"949a53fab034529867a410455cb85626abd92d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 22:22:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10abeb91ab740d7809cd5dfff84e511\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1a5fbac08b2447a88c7b307957c77eb : SUCCESS in 1h 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566b88ee2d6c41f7a8fda01f05b76bc4 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b580094b1949c88cd3b36b061685af : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed356af76377400782f2c485814d7cfe : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd90085e0b545b9a5e242659caefefd : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4383713f93408ca81addfda607158a76e31cc9c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-19 10:16:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94bf4b9dc4862b68df26dae71dc92407fdd6a7fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-19 10:16:43.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2db9444c8bbe3485c598d207edabfe257088ecdc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-19 12:11:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eb1bf4b56494c41d086c381f2a88660394f7f796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:12:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5854d5cc8bc87cbb55a66dc1c1afac683db814b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 13:24:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfa1d20c3c345829067b3a90cba0550\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25391d43e5fe4b35aa9c22d587a52f43 : SUCCESS in 1h 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53975ea64ab47e7aee6c5ff6166a4f7 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fca5785ebd5475fa0831f6defc49ec0 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90240b7897d34d3ea3a202eb4d0692c3 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a5f54e9922429f8b3df5b319b71a51 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6ca953afbfbdca5c145d2692abfe2a82751a20c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 13:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"221f4248703f49bea37c3bde41e21e89d3b2635a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 13:25:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666cfb3d979349e79afe6f0381347521\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e6532b1e4941a7a738f43f31233c6e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5189992d719ad15e0e3504947895bf5ba9dc7a1d","revisions":{"be8399bbffd5fc1ab76a54cecaa3cc7abdc789e3":{"kind":"REWORK","_number":1,"created":"2022-10-18 03:54:28.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/861695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/861695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/861695/1"}}},"commit":{"parents":[{"commit":"9eb809421d525568221ce9b6d289eedccbe5b1fc","subject":"Merge \"Add Python3 antelope unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9eb809421d525568221ce9b6d289eedccbe5b1fc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:38:37.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:54:24.000000000","tz":660},"subject":"Order old importlib-metadata results by group","message":"Order old importlib-metadata results by group\n\nA Debian Buster-based zuul-jobs test started failing when using stestr\nrecently [1].  Upon further investigation, this is a Python 3.7\nenvironment which is affected by a recent breaking change to\nimportlib_metadata.\n\nIt seems stevedore worked around this with\nIb9c2b0a14edea91e97d122d2ac93b650029f918e, which was released with\n3.5.1 -- but I was still seeing the issue.\n\nUpon further investigation, the \"real_groups\" dict being returned here\nwith importlib-metadata 4.12.0 is in buckets by group, e.g.\n\n {\u0027group.one\u0027: [EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one\u0027),\n                EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.two\u0027)],\n  \u0027group.two\u0027: [EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two\u0027),\n                EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two\u0027)],\n }\n\nThis current code seems to return a dict with entry-points by thier\nname, e.g.\n\n {\u0027foo\u0027: EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one),\n  \u0027bar\u0027: EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one),\n  \u0027moo\u0027: EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two),\n  \u0027goo\u0027: EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two)\n }\n\nThis reorgansies the fixup routine to put entry-points in a bucket by\ntheir group.  With this change, stestr is again finding it\u0027s command\nplugins.\n\n[1] https://github.com/mtreinish/stestr/issues/336\n[2] https://github.com/python/importlib_metadata/issues/409\n\nChange-Id: I3496ab1dfa312b1098a869cdfd9a0c6f81653b28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/be8399bbffd5fc1ab76a54cecaa3cc7abdc789e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/be8399bbffd5fc1ab76a54cecaa3cc7abdc789e3"}]},"branch":"refs/heads/master"},"5189992d719ad15e0e3504947895bf5ba9dc7a1d":{"kind":"REWORK","_number":2,"created":"2022-10-18 21:00:04.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/861695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/861695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/861695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/861695/2"}}},"commit":{"parents":[{"commit":"9eb809421d525568221ce9b6d289eedccbe5b1fc","subject":"Merge \"Add Python3 antelope unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9eb809421d525568221ce9b6d289eedccbe5b1fc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:38:37.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 20:57:43.000000000","tz":660},"subject":"Order old importlib-metadata results by group","message":"Order old importlib-metadata results by group\n\nA Debian Buster-based zuul-jobs test started failing when using stestr\nrecently [1].  Upon further investigation, this is a Python 3.7\nenvironment which is affected by a recent breaking change to\nimportlib_metadata.\n\nIt seems stevedore worked around this with\nIb9c2b0a14edea91e97d122d2ac93b650029f918e, which was released with\n3.5.1 -- but I was still seeing the issue.\n\nUpon further investigation, the \"real_groups\" dict being returned here\nwith importlib-metadata 4.12.0 is in buckets by group, e.g.\n\n {\u0027group.one\u0027: [EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one\u0027),\n                EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one\u0027)],\n  \u0027group.two\u0027: [EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two\u0027),\n                EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two\u0027)],\n }\n\nThis current code seems to return a dict with entry-points by thier\nname, e.g.\n\n {\u0027foo\u0027: EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one),\n  \u0027bar\u0027: EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one),\n  \u0027moo\u0027: EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two),\n  \u0027goo\u0027: EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two)\n }\n\nThis reorgansies the fixup routine to put entry-points in a bucket by\ntheir group.  With this change, stestr is again finding it\u0027s command\nplugins.\n\n[1] https://github.com/mtreinish/stestr/issues/336\n[2] https://github.com/python/importlib_metadata/issues/409\n\nChange-Id: I3496ab1dfa312b1098a869cdfd9a0c6f81653b28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/5189992d719ad15e0e3504947895bf5ba9dc7a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/5189992d719ad15e0e3504947895bf5ba9dc7a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
