)]}'
{"id":"openstack%2Fstevedore~861863","triplet_id":"openstack%2Fstevedore~stable%2Fzed~I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","project":"openstack/stevedore","branch":"stable/zed","topic":"entrypoints-py37","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-19 16:27:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","subject":"Order old importlib-metadata results by group","status":"MERGED","created":"2022-10-19 10:17:35.000000000","updated":"2022-10-19 16:28:23.000000000","submitted":"2022-10-19 16:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861863-entrypoints-py37","meta_rev_id":"bf91904903712447a9c53c7b26704d1394c1e238","_number":861863,"virtual_id_number":861863,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 16:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-19 15:04:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-19 15:04:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 10:17:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2022-10-19 12:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 15:04:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"167dbec91a6d0a8b0acac80eddc718c58f9a1c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-19 10:17:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ab5aac6fbc67f122c069fc46d3993cb9a2b4d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:01:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85bc2100005d49d5a1c27030be3e3342\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5610f2ad52db4400ab3990861d942439 : SUCCESS in 1h 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93b755a144c4ea18f4f1e4ce782938f : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a291701bd3fa463286a68fcc47ff8301 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52eb7c023f2949b597b7ad4f9677a187 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f343e39a5a2642a48b2a49efec2b80ba : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f2c819ac9546e7bbd6f8866b5eeb66 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"447b4b506acf18eaa5c5773af5a51ee9997b209a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-19 15:04:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a731be59a359409f31a2623335fb8f2e3aa0d5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:04:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef78c93552c2c5cae5115adace7f2733580c3681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:27:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deb7b4f0817a4e79b21a44ada788e2d7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c76ebcf735b24052b728ed5b20b1c3a6 : SUCCESS in 1h 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a1954c309d548c1b4600d19b4b564e1 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44c8f16be72c48b3b6b39f6e26d46d54 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32b8e86d6ce344db91db4a55a4e1f706 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450cf12d96d34ad3b086a23829bb99c7 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1bfc20d489a0a5d6294db18287495e179e8da9b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:27:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf91904903712447a9c53c7b26704d1394c1e238","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:28:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbb56de6654e4694ba437ef48b9c94da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7659d011456743a7b339458b2d1f7370 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93f1e0937c9eadd79050c12808d8a2b8b4f2646b","revisions":{"93f1e0937c9eadd79050c12808d8a2b8b4f2646b":{"kind":"REWORK","_number":1,"created":"2022-10-19 10:17:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/861863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/63/861863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/861863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/861863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/861863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/63/861863/1"}}},"commit":{"parents":[{"commit":"1e4d71d4b552494ff33ad03b76d2a590d58efe77","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1e4d71d4b552494ff33ad03b76d2a590d58efe77"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:38:37.000000000","tz":660},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-19 10:17:28.000000000","tz":60},"subject":"Order old importlib-metadata results by group","message":"Order old importlib-metadata results by group\n\nA Debian Buster-based zuul-jobs test started failing when using stestr\nrecently [1].  Upon further investigation, this is a Python 3.7\nenvironment which is affected by a recent breaking change to\nimportlib_metadata.\n\nIt seems stevedore worked around this with\nIb9c2b0a14edea91e97d122d2ac93b650029f918e, which was released with\n3.5.1 -- but I was still seeing the issue.\n\nUpon further investigation, the \"real_groups\" dict being returned here\nwith importlib-metadata 4.12.0 is in buckets by group, e.g.\n\n {\u0027group.one\u0027: [EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one\u0027),\n                EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one\u0027)],\n  \u0027group.two\u0027: [EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two\u0027),\n                EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two\u0027)],\n }\n\nThis current code seems to return a dict with entry-points by thier\nname, e.g.\n\n {\u0027foo\u0027: EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one),\n  \u0027bar\u0027: EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one),\n  \u0027moo\u0027: EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two),\n  \u0027goo\u0027: EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two)\n }\n\nThis reorgansies the fixup routine to put entry-points in a bucket by\ntheir group.  With this change, stestr is again finding it\u0027s command\nplugins.\n\n[1] https://github.com/mtreinish/stestr/issues/336\n[2] https://github.com/python/importlib_metadata/issues/409\n\nChange-Id: I3496ab1dfa312b1098a869cdfd9a0c6f81653b28\n(cherry picked from commit 5189992d719ad15e0e3504947895bf5ba9dc7a1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/93f1e0937c9eadd79050c12808d8a2b8b4f2646b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/93f1e0937c9eadd79050c12808d8a2b8b4f2646b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
