)]}'
{"id":"openstack%2Fstevedore~861865","triplet_id":"openstack%2Fstevedore~stable%2Fxena~I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","project":"openstack/stevedore","branch":"stable/xena","topic":"entrypoints-py37","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-19 16:50:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","subject":"Order old importlib-metadata results by group","status":"MERGED","created":"2022-10-19 10:18:22.000000000","updated":"2022-10-19 16:51:59.000000000","submitted":"2022-10-19 16:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861865-entrypoints-py37","meta_rev_id":"47fbb72312bf0a3c1e6d8f2494f7feccc8963f46","_number":861865,"virtual_id_number":861865,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 16:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-19 15:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-19 15:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 10:18:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2022-10-19 12:01:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 15:06:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"5c0fe2ad97b75b09a2d4dcb492b7a8f38193ea62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-19 10:18:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb806a8c825566cf7d398ef1d089dd19921222eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:01:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c544bcffd2496f8cbca5e0659ebf7a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9a55daecb98415f8aa632b4cc11485c : SUCCESS in 1h 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7138e47bb7e54d81a92a1c03f39442e4 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b18b1265531d483b92db44d29984355f : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e5dba6d6f9a405d8f644a3b163e08cc : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bddc89664bac4b7a9ec79d0997db1b63 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61170cb03ea5487dba0e5e95045848e3 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3c8dc7cfe29cedfb629561feacf1adceab133813","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-19 15:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c16a78750eec132c8d4818b2d9ee35f6ca8145f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:07:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c81ca2d9462a466cbea1aecf7cb53df822dd370b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:50:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00270f1b010a42a19ee69136e62f4f3b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/208ea79ba1e148e7afe6c8e09054cdca : SUCCESS in 1h 07m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a721a11f64549abb1373ede2c9880a7 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f311177754634e2b94627eb017fe4c1c : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4c438abed54725a4cdc8c750745584 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1e94dd85a442cabb128abab805ba4b : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2912061877289ffddd7772326121e2c691ebdf1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:50:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47fbb72312bf0a3c1e6d8f2494f7feccc8963f46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1fc87443ddb4c91b2a0ad5787ee79a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750d91c257144f16926fd0331ab33aee : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c1270666a4ec1ba164ce610b8ba798e157dacf2","revisions":{"1c1270666a4ec1ba164ce610b8ba798e157dacf2":{"kind":"REWORK","_number":1,"created":"2022-10-19 10:18:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/861865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/65/861865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/65/861865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/65/861865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/65/861865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/65/861865/1"}}},"commit":{"parents":[{"commit":"97dee6c435e4609231c96106fe253a1d9749d9cc","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/97dee6c435e4609231c96106fe253a1d9749d9cc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:38:37.000000000","tz":660},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-19 10:18:17.000000000","tz":60},"subject":"Order old importlib-metadata results by group","message":"Order old importlib-metadata results by group\n\nA Debian Buster-based zuul-jobs test started failing when using stestr\nrecently [1].  Upon further investigation, this is a Python 3.7\nenvironment which is affected by a recent breaking change to\nimportlib_metadata.\n\nIt seems stevedore worked around this with\nIb9c2b0a14edea91e97d122d2ac93b650029f918e, which was released with\n3.5.1 -- but I was still seeing the issue.\n\nUpon further investigation, the \"real_groups\" dict being returned here\nwith importlib-metadata 4.12.0 is in buckets by group, e.g.\n\n {\u0027group.one\u0027: [EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one\u0027),\n                EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one\u0027)],\n  \u0027group.two\u0027: [EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two\u0027),\n                EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two\u0027)],\n }\n\nThis current code seems to return a dict with entry-points by thier\nname, e.g.\n\n {\u0027foo\u0027: EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one),\n  \u0027bar\u0027: EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one),\n  \u0027moo\u0027: EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two),\n  \u0027goo\u0027: EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two)\n }\n\nThis reorgansies the fixup routine to put entry-points in a bucket by\ntheir group.  With this change, stestr is again finding it\u0027s command\nplugins.\n\n[1] https://github.com/mtreinish/stestr/issues/336\n[2] https://github.com/python/importlib_metadata/issues/409\n\nChange-Id: I3496ab1dfa312b1098a869cdfd9a0c6f81653b28\n(cherry picked from commit 5189992d719ad15e0e3504947895bf5ba9dc7a1d)\n(cherry picked from commit 93f1e0937c9eadd79050c12808d8a2b8b4f2646b)\n(cherry picked from commit 6c9978a9062da6cc3f9cc1859bf63c72b057cabb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1c1270666a4ec1ba164ce610b8ba798e157dacf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1c1270666a4ec1ba164ce610b8ba798e157dacf2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
