)]}'
{"id":"openstack%2Fstevedore~861866","triplet_id":"openstack%2Fstevedore~stable%2Fwallaby~I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","project":"openstack/stevedore","branch":"stable/wallaby","topic":"entrypoints-py37","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-19 17:13:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3496ab1dfa312b1098a869cdfd9a0c6f81653b28","subject":"Order old importlib-metadata results by group","status":"MERGED","created":"2022-10-19 10:18:45.000000000","updated":"2022-10-19 17:14:12.000000000","submitted":"2022-10-19 17:13:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861866-entrypoints-py37","meta_rev_id":"83a3ece3e0744c26fca95928b83cc5c1f5e338b2","_number":861866,"virtual_id_number":861866,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 17:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-10-19 15:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-10-19 15:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 10:18:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2022-10-19 11:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 15:07:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"6d75282009a6e11e96c346e804e3fb0930e3c29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-19 10:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46a2920e1d51adf91f43f5ca7cb9eac5a29088a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72b8618d740e4c61b01aa1dcae0e37a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4231a14dff4643879082a75642e5886a : SUCCESS in 1h 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652206c48ad840d7981cc95bcecfbedd : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03915a55149d4d62846ed1f69764c8fe : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e39c743e182146039c3ec0078d1d09ae : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e9390455d04aafab2d63e0c5875dbe : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af868c3425cf4eb79cb75ad52ccb2b51 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ec3d681392532950b433f23720f9e61aa5336543","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-19 15:07:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b9956334cb472994dcce52e8c599b3fb553de49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:07:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0661da90e4c948354df4e255f3dc75c3774c4579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:13:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e9bf8b1a9f4bf0959910dd17b5af5a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de3e5705726841adb417ee67a6514fce : SUCCESS in 2h 00m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29069ec6d4504d58b75d9dd790048ba5 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa10691c0434497bab7516da2c5f9ad : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d985e1307114a708d17418a6a7aa63d : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64366567e9754a0d8c037cf5e91da959 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a73831508d5765309cb7c00c544fdae20e0647d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:13:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83a3ece3e0744c26fca95928b83cc5c1f5e338b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:14:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0468f50816c4b54a0e30fe7b8194bb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f379e2f81046da801d0d7a38f7604c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b48fffd01307223b53ef129bb6a0753a67abdce","revisions":{"7b48fffd01307223b53ef129bb6a0753a67abdce":{"kind":"REWORK","_number":1,"created":"2022-10-19 10:18:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/861866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/66/861866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/66/861866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/66/861866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/66/861866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/66/861866/1"}}},"commit":{"parents":[{"commit":"48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340","subject":"Fix compatibility with Python 3.12, importlib-metadata 5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/48e178ab5ca6d8d2e6c6b9eee9e2fc62176d1340"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-10-18 03:38:37.000000000","tz":660},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-10-19 10:18:40.000000000","tz":60},"subject":"Order old importlib-metadata results by group","message":"Order old importlib-metadata results by group\n\nA Debian Buster-based zuul-jobs test started failing when using stestr\nrecently [1].  Upon further investigation, this is a Python 3.7\nenvironment which is affected by a recent breaking change to\nimportlib_metadata.\n\nIt seems stevedore worked around this with\nIb9c2b0a14edea91e97d122d2ac93b650029f918e, which was released with\n3.5.1 -- but I was still seeing the issue.\n\nUpon further investigation, the \"real_groups\" dict being returned here\nwith importlib-metadata 4.12.0 is in buckets by group, e.g.\n\n {\u0027group.one\u0027: [EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one\u0027),\n                EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one\u0027)],\n  \u0027group.two\u0027: [EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two\u0027),\n                EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two\u0027)],\n }\n\nThis current code seems to return a dict with entry-points by thier\nname, e.g.\n\n {\u0027foo\u0027: EntryPoint(name\u003d\u0027foo\u0027, ... , group\u003d\u0027group.one),\n  \u0027bar\u0027: EntryPoint(name\u003d\u0027bar\u0027, ... , group\u003d\u0027group.one),\n  \u0027moo\u0027: EntryPoint(name\u003d\u0027moo\u0027, ... , group\u003d\u0027group.two),\n  \u0027goo\u0027: EntryPoint(name\u003d\u0027goo\u0027, ... , group\u003d\u0027group.two)\n }\n\nThis reorgansies the fixup routine to put entry-points in a bucket by\ntheir group.  With this change, stestr is again finding it\u0027s command\nplugins.\n\n[1] https://github.com/mtreinish/stestr/issues/336\n[2] https://github.com/python/importlib_metadata/issues/409\n\nChange-Id: I3496ab1dfa312b1098a869cdfd9a0c6f81653b28\n(cherry picked from commit 5189992d719ad15e0e3504947895bf5ba9dc7a1d)\n(cherry picked from commit 93f1e0937c9eadd79050c12808d8a2b8b4f2646b)\n(cherry picked from commit 6c9978a9062da6cc3f9cc1859bf63c72b057cabb)\n(cherry picked from commit 1c1270666a4ec1ba164ce610b8ba798e157dacf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7b48fffd01307223b53ef129bb6a0753a67abdce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7b48fffd01307223b53ef129bb6a0753a67abdce"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
