)]}'
{"id":"openstack%2Fstevedore~882195","triplet_id":"openstack%2Fstevedore~master~Ibf0960703d78ccfcb39421ee4b1cfee2c787d85a","project":"openstack/stevedore","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-08 09:18:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-08 11:15:48.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-08 11:15:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf0960703d78ccfcb39421ee4b1cfee2c787d85a","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","status":"MERGED","created":"2023-05-03 20:01:59.000000000","updated":"2023-05-08 11:16:52.000000000","submitted":"2023-05-08 11:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":879931,"submission_id":"882195-oslo-drop-py38","meta_rev_id":"b38b9e376e7c520f3421ffa096339ef1cc1d627b","_number":882195,"virtual_id_number":882195,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-08 11:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-05-08 09:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-05-08 09:18:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-03 20:01:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-05-03 20:01:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-03 20:01:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cd51a2300b999a1bd0f1e94254623754367205f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bc3951de28779d30cfbdc59de6e63777a1186dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 21:25:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d7dc4db29c4cbf912272703f63621e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2bf85daf2fa445f8ae0ca3deeb26a904 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b311e53b990441f9de5568e9e353687 : SUCCESS in 1h 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b92cf4047b3471b8ee7096f35c64b82 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f27b170a0146959b65323170974440 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58d0a6f24c59449893aa1ec595dc413f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b82ed841ef24c0facb14bd512083cc7 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"80c201a6c767ccc062fda29e5e830bcaf8e61bcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-04 20:20:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"80cbbc8d481518eb19233d3f79cd4058b8ea1137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:42:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ab773770534e12ab6cf5109c901a22\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af8747943f0844ce87c27ad0d80918ad : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/218ae7d779544eada6a6ae111f18fa55 : SUCCESS in 1h 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50bda241f3f41a68e491053952511b6 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4556026aecb84c2e86885d010f2fa930 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b3484cdff324623b3a54273d7726ac1 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637e6d323fc9485caa2375280029e619 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"71a66ef89e4085dbb2fc9676d6c11cc32ff9457b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-05 12:08:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cabf96c612463ed9bd8553376e5517f2aad81660","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-05 18:50:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec6591b59fd34c3c219ee5e02aaf5d71540949c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-05 18:50:50.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf8f35dc52f51e895700ac321a15ecf01ec9d96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 20:21:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f774f4360544b40919785f25d4f388c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47a2586598ba40fdbefa513010631b37 : SUCCESS in 1m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e45e88297984808bbb469c40f4ce0d6 : SUCCESS in 1h 25m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29498657363543e389023e936af82519 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/186305969e1449f4a2ea32833c596e8e : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e530575373c9446983e0bc9b980764e2 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33061924f8da4eadaa23e83fb9d40a34 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"38bf0bac92fde524c6ad1dc19a6bfdae837b3bc4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-08 09:18:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"69389e02ab4a989f55f79e18c0a5929ce186d08f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 09:18:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ebf61de8535dbe592d3208af27cd6e492b8a8f78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 11:15:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb6147e344c04a27afee3c1f1bff6c27\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4904984b76b94f159530a87e4bb234f5 : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f943eb528f6d4723abdcae7df767a849 : SUCCESS in 1h 46m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b465664a4df5457590e08372b193c1df : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42745abfe22443be84d98b7056ed6c92 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ae23ef9a9d24de1a15e1a6c2d604099 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b9a6cb993d4c81af6b8c53f3299c2c : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"411ef443394fe6bf8fefb56a97ca02100d5ddd2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 11:15:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b38b9e376e7c520f3421ffa096339ef1cc1d627b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 11:16:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe73e81a0af4bf5bf90c383d060039b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6050e74d79f489d90a769fe42814e42 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc8f04e5f08d67ebf3a7f4508f37320319c40edc","revisions":{"83039ff4430315ab5273d9d65ce042d270687e77":{"kind":"REWORK","_number":1,"created":"2023-05-03 20:01:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/882195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/882195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/882195/1"}}},"commit":{"parents":[{"commit":"da129b1c53b8f529656c48a26d495a8664a3c901","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/da129b1c53b8f529656c48a26d495a8664a3c901"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:59.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:59.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit da129b1c53b8f529656c48a26d495a8664a3c901.\n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: Ibf0960703d78ccfcb39421ee4b1cfee2c787d85a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/83039ff4430315ab5273d9d65ce042d270687e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/83039ff4430315ab5273d9d65ce042d270687e77"}]},"branch":"refs/heads/master"},"0465398cd907fe434ae4d3d1baef6e5c7f4cfc61":{"kind":"REWORK","_number":2,"created":"2023-05-04 20:20:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/882195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/882195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/882195/2"}}},"commit":{"parents":[{"commit":"da129b1c53b8f529656c48a26d495a8664a3c901","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/da129b1c53b8f529656c48a26d495a8664a3c901"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:59.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:20:17.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit da129b1c53b8f529656c48a26d495a8664a3c901.\n\nKeeping Python 3.10 in setup.cfg classifier. \n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: Ibf0960703d78ccfcb39421ee4b1cfee2c787d85a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0465398cd907fe434ae4d3d1baef6e5c7f4cfc61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0465398cd907fe434ae4d3d1baef6e5c7f4cfc61"}]},"branch":"refs/heads/master"},"cc8f04e5f08d67ebf3a7f4508f37320319c40edc":{"kind":"REWORK","_number":3,"created":"2023-05-05 18:50:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/882195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/95/882195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/95/882195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/95/882195/3"}}},"commit":{"parents":[{"commit":"da129b1c53b8f529656c48a26d495a8664a3c901","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/da129b1c53b8f529656c48a26d495a8664a3c901"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:59.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-05 18:50:48.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit da129b1c53b8f529656c48a26d495a8664a3c901.\n\nKeeping Python 3.10 in setup.cfg classifier and zuul.yaml changes. \n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: Ibf0960703d78ccfcb39421ee4b1cfee2c787d85a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/cc8f04e5f08d67ebf3a7f4508f37320319c40edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/cc8f04e5f08d67ebf3a7f4508f37320319c40edc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
