)]}'
{"id":"openstack%2Fstevedore~918252","triplet_id":"openstack%2Fstevedore~master~I005244693e84a24f16fb7f42ec37ea2847967ad8","project":"openstack/stevedore","branch":"master","topic":"reno-eom-zed","hashtags":[],"change_id":"I005244693e84a24f16fb7f42ec37ea2847967ad8","subject":"reno: Update master for unmaintained/zed","status":"MERGED","created":"2024-05-06 05:58:29.000000000","updated":"2024-05-06 11:28:29.000000000","submitted":"2024-05-06 11:27:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918252-reno-eom-zed","meta_rev_id":"5b034451154d72fa9f393e4969137241a1824b5f","_number":918252,"virtual_id_number":918252,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-06 11:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-05-06 11:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-05-06 11:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-06 06:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 11:14:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"2ca7c8d0ca92ea1fe9b0c8709d2da12ff8ba864d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-05-06 05:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"267944b0177e820220e408395d4310d3e079de90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 06:31:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c85b90a917a4761aa04cc046df5232c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16f254e53d74343b53302c0b07aa764 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec63e45718e2414589fa7a205faf208e : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53f0ae9b56ac42848f3d14151a009f21 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ab4ab79a36dc00175752e06736bfe973e949bf8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-06 11:14:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4319208af1eace94c56e965cafdb228b08152a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:14:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a62f950048e23f4042dfea546d1473f4b8bca925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:27:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e5bc526d4384a2a9df18b68c35e46e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30e4d6db65b47eca7676d2695e71923 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f930db859b8c43cab23d8d7eea80d683 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/864ed32070d24d8092d92a1f9be01477 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2a5f3c10b659902980cce34fc932ff3374ecd634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:27:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b034451154d72fa9f393e4969137241a1824b5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:28:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7224cc171c4c838738edd78220421e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314d40cf90fd463ab792c2f4520bf952 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0213711df33c4fd68c09f1ad10e7b638 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51134a4dc04872df9f736e5ae52d4b665c40e364","revisions":{"51134a4dc04872df9f736e5ae52d4b665c40e364":{"kind":"REWORK","_number":1,"created":"2024-05-06 05:58:29.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/52/918252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/52/918252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/52/918252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/52/918252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/52/918252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/52/918252/1"}}},"commit":{"parents":[{"commit":"8ab92449ad413dc7593218d1c05c36327f1b06c3","subject":"Remove old excludes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/8ab92449ad413dc7593218d1c05c36327f1b06c3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-06 05:58:29.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-06 05:58:29.000000000","tz":0},"subject":"reno: Update master for unmaintained/zed","message":"reno: Update master for unmaintained/zed\n\nUpdate the zed release notes configuration to build from\nunmaintained/zed.\n\nChange-Id: I005244693e84a24f16fb7f42ec37ea2847967ad8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/51134a4dc04872df9f736e5ae52d4b665c40e364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/51134a4dc04872df9f736e5ae52d4b665c40e364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
