)]}'
{"id":"openstack%2Fstevedore~928251","triplet_id":"openstack%2Fstevedore~master~Ib7fca8a441635f79d2aa39a81dc85ae6ce4c44df","project":"openstack/stevedore","branch":"master","topic":"reno-2024.2","hashtags":[],"change_id":"Ib7fca8a441635f79d2aa39a81dc85ae6ce4c44df","subject":"Update master for stable/2024.2","status":"MERGED","created":"2024-09-05 16:03:06.000000000","updated":"2024-09-12 10:24:33.000000000","submitted":"2024-09-12 10:23:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928251-reno-2024.2","meta_rev_id":"492e3fada61c4e9c6bfde41c44e610512adfcd9c","_number":928251,"virtual_id_number":928251,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 10:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-09-12 10:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-09-12 10:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-05 16:18:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-12 10:15:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"08b1264463285a94f4d423e0836fda9c1b5a6658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-09-05 16:03:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c465dc82b2335732e0b2e5960aea86beb0b49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 16:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb08dcef21e44afb21c4f60dfdc3d72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4409f3578b48d6bca477e8a57bb123 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83cdfe8003c4e3891d5950ec44981d7 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a9a0a5784748a5ae58502d9d698810 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"09c100e91a0d732c782294a001a6eea48abb1e31","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 10:15:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c12207705666d724a427c5860d9a4f2484134c95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:15:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf1dfe0611efcfffa5a565f75f5a349d45b21454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:23:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff7760148124ab18c979786a6958bef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a80063d04a47efa215161eb1ba4322 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08813e5a514346c08572fa9f82787e0f : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955617bbd1344ecc97048e37796737af : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d31db88ee1f9f73b29c370ff6b47b071cbffc752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:23:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"492e3fada61c4e9c6bfde41c44e610512adfcd9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:24:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afe320bc50364475964a80e321512d0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bc834aaee540e698bd474dbde47d88 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2125f1265e824b48902ac5f29687ca63 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49324f8e6d38e47ff27342e49d5c43419336ecd3","revisions":{"49324f8e6d38e47ff27342e49d5c43419336ecd3":{"kind":"REWORK","_number":1,"created":"2024-09-05 16:03:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/51/928251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/51/928251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/928251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/928251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/928251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/51/928251/1"}}},"commit":{"parents":[{"commit":"51134a4dc04872df9f736e5ae52d4b665c40e364","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/51134a4dc04872df9f736e5ae52d4b665c40e364"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-05 16:03:06.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-05 16:03:06.000000000","tz":0},"subject":"Update master for stable/2024.2","message":"Update master for stable/2024.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2024.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2024.2.\n\nSem-Ver: feature\nChange-Id: Ib7fca8a441635f79d2aa39a81dc85ae6ce4c44df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/49324f8e6d38e47ff27342e49d5c43419336ecd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/49324f8e6d38e47ff27342e49d5c43419336ecd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
