)]}'
{"id":"openstack%2Fstevedore~957398","triplet_id":"openstack%2Fstevedore~master~Id6d5c86b8975030b3cf5894de94f474b4a7ea829","project":"openstack/stevedore","branch":"master","topic":"release-health-check-flamingo","hashtags":[],"change_id":"Id6d5c86b8975030b3cf5894de94f474b4a7ea829","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:50:44.000000000","updated":"2025-08-15 07:53:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97c4766ba47e1781ac33e7e2ecc13f2612b67e6d","_number":957398,"virtual_id_number":957398,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-15 03:18:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-15 03:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b09a8679e620857a8a292190f2ca2d4935941461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cd01d7c10004a86d5a9280362f8450d0472ba19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 03:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5c580af01d4e04835c02b0bc25cb86\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c516748f34064d00812bc59cde07db12 : SUCCESS in 56m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae0dd71145248b2af5dfe84c7369575 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a665f582c548d4bd8bfb11b710fbd2 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/667b3f65b45646c7abb6f6a60211170c : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78b11af4b53c459baef9deef33862ea4 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887248d0eb5a4a05b3a4ea994449d110 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"97c4766ba47e1781ac33e7e2ecc13f2612b67e6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-15 07:53:55.000000000","message":"Abandoned\n\ngate jobs are passing, patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ecafb742fc4a3ce31ce7513a7d51eccb8750b88","revisions":{"1ecafb742fc4a3ce31ce7513a7d51eccb8750b88":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:50:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/957398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/98/957398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/957398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/957398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/98/957398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/98/957398/1"}}},"commit":{"parents":[{"commit":"3c04bde90db9e6d1ab459ad507b48f3894e6832c","subject":"Merge \"Remove explicit pbr dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/3c04bde90db9e6d1ab459ad507b48f3894e6832c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:50:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:50:43.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Id6d5c86b8975030b3cf5894de94f474b4a7ea829\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1ecafb742fc4a3ce31ce7513a7d51eccb8750b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1ecafb742fc4a3ce31ce7513a7d51eccb8750b88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
