)]}'
{"id":"openstack%2Fstevedore~969055","triplet_id":"openstack%2Fstevedore~master~I3011f918ebcb0d25c80f40d6a95025bb97f4e6b5","project":"openstack/stevedore","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-05 17:24:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I3011f918ebcb0d25c80f40d6a95025bb97f4e6b5","subject":"Run mypy from tox","status":"MERGED","created":"2025-12-01 13:33:34.000000000","updated":"2025-12-05 18:24:59.000000000","submitted":"2025-12-05 18:24:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969055-typing","meta_rev_id":"d9a6aea297ef216ec3f3017c76f83cf5678f3c49","_number":969055,"virtual_id_number":969055,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-05 17:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-05 17:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 15:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-02 15:55:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"8362bfc4509bd3dd9a4b3141795028aaf1bd1322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 13:33:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18ae9afa2f9a4157ed1aa73fe2d68a145cb6edbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 15:20:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5875813f2934caa90c1bfcea38add89\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d758c967c0894e188b87a269b186f102 : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90d97ddbce174fb88f29054fcbe56d3d : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/940891dbe18f426a84cc9b16c73cb23c : SUCCESS in 1h 42m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220232fb083444d7ba35af01bfc118b3 : FAILURE in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a14f6b70b54b2bafb21ba9dcff402d : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00731a00eaf4420392fd5a62d78ba74c : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce14569d50643488796317945046912 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/758d750bf1014ce1ad1628dd2eb93777 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6d76b7beb2c37a30ad916fb8003dc616ad5e92d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 16:43:07.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"04ec02f00109aa36f8cdcdb21884fb9f48bca7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 17:36:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226c26f782f3491a9f7126d00af66cd5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/668fe8d0788b4822928b074725b70716 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/87cc1b12e6994aab847883cb5e14b95d : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91585d53676c4a16be88f0dc0737b4d1 : SUCCESS in 51m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d80c9b98a814c9cbf5443a13fc0291a : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f634c331fd0d4089af1f8ca8506e05ff : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb3e291881d94f929b217829993afb41 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d65ccfa36f240bd88bc364068c4f543 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4a0a2b5976f43b48f7a3bd915061f5f : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a192fd368d04899a12333f89b4cb995628eef4ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-02 15:55:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c1fbf3639c63f39d33ede57a495a215043f9ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:15:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e3212edf2c2bf4cc366c1c21fd4d3808d9ba6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 20:07:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f995a937b0324b7db58e7f3ca37c64ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a88a9f027d74272a1f9d0dd12ac6cfe : SUCCESS in 1m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfb4b6431373452eaaf51e67afaf2f7e : SUCCESS in 2m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f3c7aca1e0a4cb4aa72ccf638db7e44 : SUCCESS in 1h 47m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e049f26c3f474f41917e76a269a6ae62 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae0ed5d26f5a4ffd80eb4e56c4eb1379 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af35e729856a42bc8bb5658b9326f159 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fc7380ad52479cb1ee5c41eb70e11b : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c260bd3dffd948aeb68e354c23cafefb : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"22d8b66f9927cbc994c9b402b60c905630391dda","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-05 17:24:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e7fd9fa0ed26f732cf135d95b5909baabf498210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:25:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb66be18f8253d1e2b30219324c20ff46bdfe4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:24:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a65948c0f347658fc775ebe658c1a5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ab47305d8bf43f38335376a8939c087 : SUCCESS in 2m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/50d8c966fad94237a2c804c08541f226 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c386328e9b4475ea6f2ce4b8706d04b : SUCCESS in 50m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b64e7eda534f00aa460932135830f5 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50b3accd979a4c259b926a1fbe6a4288 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dea8f08311348638e76d06543e18c4b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4840cad3833645c8a2faddbf389fca34 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0468b0b6f21d49bf99a140c089b5dec4 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ea265e2b205f4540cc23bb79f032c87cbda5fa1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:24:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d9a6aea297ef216ec3f3017c76f83cf5678f3c49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:24:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aeb7f8dd06444b291576d9b1b790d1a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9422bd15c2f44b1ba4c797d667e89d95 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ad23d0e88554bacb6559052575f7eef : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbb2ed14957aadc56fe9eba6ed68e7f92079102b","revisions":{"7260a4ee12b68526e527aceb59bfaef2b953e0e5":{"kind":"REWORK","_number":1,"created":"2025-12-01 13:33:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/969055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/55/969055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/55/969055/1"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 12:59:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 13:33:29.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3011f918ebcb0d25c80f40d6a95025bb97f4e6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7260a4ee12b68526e527aceb59bfaef2b953e0e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/7260a4ee12b68526e527aceb59bfaef2b953e0e5"}]},"branch":"refs/heads/master"},"1fe6c74571070a5861c2c163393bda96c96a1a66":{"kind":"REWORK","_number":2,"created":"2025-12-01 16:43:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/969055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/55/969055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/55/969055/2"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 12:59:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 16:43:07.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3011f918ebcb0d25c80f40d6a95025bb97f4e6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1fe6c74571070a5861c2c163393bda96c96a1a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1fe6c74571070a5861c2c163393bda96c96a1a66"}]},"branch":"refs/heads/master"},"bbb2ed14957aadc56fe9eba6ed68e7f92079102b":{"kind":"REWORK","_number":3,"created":"2025-12-04 18:15:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/969055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/55/969055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/55/969055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/55/969055/3"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 12:59:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-04 18:14:59.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3011f918ebcb0d25c80f40d6a95025bb97f4e6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/bbb2ed14957aadc56fe9eba6ed68e7f92079102b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/bbb2ed14957aadc56fe9eba6ed68e7f92079102b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
