)]}'
{"id":"openstack%2Fstevedore~969463","triplet_id":"openstack%2Fstevedore~master~I1cc4ed756f044c2bb564e091c6474b2894583bed","project":"openstack/stevedore","branch":"master","topic":"H904","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-04 19:36:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1cc4ed756f044c2bb564e091c6474b2894583bed","subject":"Delay string interpolations at logging calls","status":"MERGED","created":"2025-12-03 14:14:31.000000000","updated":"2025-12-04 19:37:07.000000000","submitted":"2025-12-04 19:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969463-H904","meta_rev_id":"8397db57a5f5d5bb9a1a093b8bf1dc25ad1c9a92","_number":969463,"virtual_id_number":969463,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 19:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-04 18:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-12-04 16:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-04 18:00:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 14:30:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 16:02:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-12-04 18:00:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"4a55e868e47020880a1ce755c308820ad1746b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 14:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31f658a65e590f11f718a6cb3c17999323ceb593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 14:30:58.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e7f3e193c194d8693cf5353ae2724b2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74441cfce5fc4023b4865ebca1addc09 : FAILURE in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936c22fe376e4005865d59588089efe5 : FAILURE in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/568700e149564797b0bce02060543258 : FAILURE in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12da2f89326b441d9a28f47f3ea95e89 : FAILURE in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9242192e90d246c99479d5b2ce4438b3 : FAILURE in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaa1ed7aabc54c5db391357e7bd17720 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e2e22f261076ef21f6c9174aa4b60f83e32dd2e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 14:35:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"83fba212a13ef560052a7d159c399d6394fe4b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 14:36:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f161057e96bbb6a803f0890067f4e340df3ccbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 16:05:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12dee3725eef4ef4a3930cb1b3cec570\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db764a00623a40329908c9076525dcc1 : SUCCESS in 1h 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ca1a20aa504012a87dab28d96c0d16 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a703b333bd5447992753ac4bf6b8cc2 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10aba237c8d044b2aa892064aadea874 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d303bb65c29140c2b56d8621a9c78cd6 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1ff3c1b84747d1abfcc1d6290a741e : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2bf8999e87387cdfe669ebc9897ea7226376cead","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-04 15:06:41.000000000","message":"Topic set to H904","accounts_in_message":[],"_revision_number":3},{"id":"be94f9df7dcea9f880fe0845e255adacaec956b7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-12-04 16:02:43.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6226e101ce3d59269a486cb8ff19e1c36dbda1a0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:00:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ab1deac39a726be66a461a4c2a88273d9fae2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:01:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d80ad92f5e62f2a6da478eba7b4573ac919fa1d3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:02:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d759799f9ba923b98e9f04f8ece4a27a56ea6e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 19:36:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad889c27c448499695dec0de9581112e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84d53c5ed22b46a8a7fb57a4a0e8e57e : SUCCESS in 1h 30m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4c62db8bb240e3be798a90cb2dae2a : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c84d98f407c84697918c9a6c02601f45 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d107dfd8c77c409fb5d11b2a4c78f8d3 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b611c60a9a7442b4a8dfc39e1c40f133 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac1040cf253444f08d87f9b5dba8e689 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"7125fc0b5119fb81beba4e07c81efcdb6238ec8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 19:36:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8397db57a5f5d5bb9a1a093b8bf1dc25ad1c9a92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 19:37:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68da24c83574a75924c54caaa517a26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c99c1452634c498ec599290db07ac4 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab744b330f9e40cba27346edd5a1fea5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee5cead3108e5c1eca17d3a56a21545f8ef1996d","revisions":{"a1e87b934f9d993c89b3064b9639fc24dbcb6f7d":{"kind":"REWORK","_number":1,"created":"2025-12-03 14:14:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/969463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/63/969463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/63/969463/1"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:14:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:14:20.000000000","tz":540},"subject":"Delay string interpolations at logging calls","message":"Delay string interpolations at logging calls\n\nThis is a general best practise which brings (slight) performance\nimprovements and also structured logging.\n\nEnable the hacking check to enforce it.\n\nChange-Id: I1cc4ed756f044c2bb564e091c6474b2894583bed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a1e87b934f9d993c89b3064b9639fc24dbcb6f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/a1e87b934f9d993c89b3064b9639fc24dbcb6f7d"}]},"branch":"refs/heads/master"},"fe6dd64d03f17f21a30b220eb83ace438b18af86":{"kind":"REWORK","_number":2,"created":"2025-12-03 14:35:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/969463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/63/969463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/63/969463/2"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:14:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:35:39.000000000","tz":0},"subject":"Delay string interpolations at logging calls","message":"Delay string interpolations at logging calls\n\nThis is a general best practise which brings (slight) performance\nimprovements and also structured logging.\n\nEnable the hacking check to enforce it.\n\nChange-Id: I1cc4ed756f044c2bb564e091c6474b2894583bed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fe6dd64d03f17f21a30b220eb83ace438b18af86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fe6dd64d03f17f21a30b220eb83ace438b18af86"}]},"branch":"refs/heads/master"},"ee5cead3108e5c1eca17d3a56a21545f8ef1996d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-03 14:36:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/969463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/63/969463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/63/969463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/63/969463/3"}}},"commit":{"parents":[{"commit":"1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/1b97a87e0051bd7bc52ad9e0b608aaf6c02db70e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:14:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 14:36:03.000000000","tz":0},"subject":"Delay string interpolations at logging calls","message":"Delay string interpolations at logging calls\n\nThis is a general best practice which brings (slight) performance\nimprovements and also structured logging.\n\nEnable the hacking check to enforce it.\n\nChange-Id: I1cc4ed756f044c2bb564e091c6474b2894583bed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ee5cead3108e5c1eca17d3a56a21545f8ef1996d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/ee5cead3108e5c1eca17d3a56a21545f8ef1996d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
