)]}'
{"id":"openstack%2Fstevedore~989013","triplet_id":"openstack%2Fstevedore~master~I27927aab7c23c9e3d15782441a29a5697625dce9","project":"openstack/stevedore","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-28 09:07:57.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-05-28 07:26:22.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I27927aab7c23c9e3d15782441a29a5697625dce9","subject":"Drop support for Python 3.10","status":"MERGED","created":"2026-05-18 17:15:56.000000000","updated":"2026-05-28 09:08:57.000000000","submitted":"2026-05-28 09:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989013","meta_rev_id":"c24ecd071e41e8795929944c2a80a8b1fe803a2b","_number":989013,"virtual_id_number":989013,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 09:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-21 16:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-05-28 07:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-05-28 07:26:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 18:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-20 07:54:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2026-05-21 16:12:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"7a499c9abd0c7f8f698c694da53933a595dd5e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-18 17:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ace0be0d70a2c30edfab69f5eb0f40f6278c43a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 18:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4230668728c48dcb21f40ffd2d47785\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1213c2e5e534e8788e41ce7a38293eb : SUCCESS in 3m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4af0257e337b40d39057ec9f65bf0677 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cb2487f9e5d4c4e954d446932bf9a44 : SUCCESS in 55m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a245edc72ff64b7e857c2ac2c4d3eaaa : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/766ea5e5678144359e40c13247602f0b : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba0e751de98746269ea4905e0c539cbc : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4777f969cdb04cdf9992d96049b0d81f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e1fcb83a4a4e5dab8c0aa064c4d935 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7de856e30749f3a8957d46efc8fa4a : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bcf2279089b25f37580f5d951d17b58974cf823f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-20 07:54:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba7908ea596774fd765d7ffc37f00d5e926b95c1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 13:13:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a8d20a45f717b1dc693ebe5c0f728f42352fe5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 13:14:37.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5c4a111d69a2188f46256ec2f454c59946e0ff53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:31:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6cdb7f0d604a03bb10dc384cc0d37e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c6b954159a744009a4cff0feeb273b4 : SUCCESS in 1m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a7105e365cd4e689130e44cc0c3942e : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/667e3359bf8f4a28ac0d4e11bc3ed5ff : SUCCESS in 2h 07m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7761b43128eb4c14a2163b7ff6ca82fa : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef1af3c8e18b434893140311a76a6974 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b20b8403fd149819f6aec8cf223f8ce : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49dfc92ba1d148b0ae88c3a1c988ba38 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27dff2157b8f4023aed0f57e097afa56 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/607ac0fd6765401e895b84284786be18 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a48ec802d1f89a614b6aa25e5abc98a6e4ecfa27","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-21 16:12:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e76890b033a63bed11a9763357f279d238174c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-28 07:26:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74438aaa776e89c58102dcbcd0a6f55b5b02c5bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 07:26:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c31bbf3112fe72acd3e9381fe5d415a7e749b7a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:07:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c32763607c2402f98959cd5090980b4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11af83cc1b374ba8a008023f32d4a5f6 : SUCCESS in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90d38e024099445c8583902320f53658 : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00a8c4b56d5e4fc884d147ae7985b717 : SUCCESS in 1h 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe1b84482ab42c0b74c44fe40c2879d : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/419c9074979848ec86c720a38de9d76d : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5aab41d25e1b4e6eb8f73829d21346e2 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723f5e2896d349ef84c7f9b97ddac289 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f32af6e46b4a0d9a11c983a6a449e2 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7e700d6b8b282680417a7348fae3cead5b6d6247","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:07:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c24ecd071e41e8795929944c2a80a8b1fe803a2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:08:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecaa8c51faab49d89103d90483abff55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6161cc82d7334f5c9d58be3e071d1cfc : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20227eda87d840b5b15ab5d3cdce37ae : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"509f705f755aa8c96a3461b40def518dc8893c52","revisions":{"fea877177d2b14f691dff2007985c97e6fd9cad1":{"kind":"REWORK","_number":1,"created":"2026-05-18 17:15:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/989013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/13/989013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/13/989013/1"}}},"commit":{"parents":[{"commit":"0b93f90bf601361d2cabe3d915f5c116695eac00","subject":"Merge \"typing: Correct type of Extension.plugin argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0b93f90bf601361d2cabe3d915f5c116695eac00"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-18 17:15:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-18 17:15:27.000000000","tz":60},"subject":"Drop support for Python 3.10","message":"Drop support for Python 3.10\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I27927aab7c23c9e3d15782441a29a5697625dce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fea877177d2b14f691dff2007985c97e6fd9cad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fea877177d2b14f691dff2007985c97e6fd9cad1"}]},"branch":"refs/heads/master"},"509f705f755aa8c96a3461b40def518dc8893c52":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-20 13:14:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/989013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/13/989013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/13/989013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/13/989013/2"}}},"commit":{"parents":[{"commit":"0b93f90bf601361d2cabe3d915f5c116695eac00","subject":"Merge \"typing: Correct type of Extension.plugin argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0b93f90bf601361d2cabe3d915f5c116695eac00"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-18 17:15:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 13:14:37.000000000","tz":0},"subject":"Drop support for Python 3.10","message":"Drop support for Python 3.10\n\nNote that ruff insists we drop use of `typing_extensions` since\nthe objects used from that library are now provided by stdlib\n`typing`.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I27927aab7c23c9e3d15782441a29a5697625dce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/509f705f755aa8c96a3461b40def518dc8893c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/509f705f755aa8c96a3461b40def518dc8893c52"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
