)]}'
{"id":"openstack%2Fstorlets~1004674","triplet_id":"openstack%2Fstorlets~master~I5297d2ce4391b40369d4efa1874f9b95f8bda227","project":"openstack/storlets","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-08 18:57:53.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I5297d2ce4391b40369d4efa1874f9b95f8bda227","subject":"Replace scripts by module executions","status":"MERGED","created":"2026-09-08 17:06:26.000000000","updated":"2026-09-08 19:15:14.000000000","submitted":"2026-09-08 19:14:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004674","meta_rev_id":"15c170df292ed1be73d7de29dec863b046e4f2ff","_number":1004674,"virtual_id_number":1004674,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 19:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-08 18:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-08 18:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 17:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c046503c834cdc1355c52f8a72f4e1167ae7e51f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 17:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91b79d2f43ae3fdf148fee228c85f1c546c50433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:38:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bb06df6f6f47babc50e2ab84dc2c7b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a94da4d8a7e48af844f500563d12ee1 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5d009a598344298532fa918d8bc8c6 : FAILURE in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08107167a801498796a8d6f46de24cee : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fd2df466f81476bafe559163aba58a5 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38ffbf85bd80429fae8424dab815ffe0 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afde3949c69b411b949e5e07d9addaf4 : SUCCESS in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c3d2b50e6904b9e85aadc219e955945 : SUCCESS in 3m 49s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/1f855e4b89c34a7581d6e702cb429cf6 : FAILURE in 30m 11s\n- storlets-functional-openjdk21 https://zuul.opendev.org/t/openstack/build/1d270de87c8540de82248d67471adb71 : FAILURE in 14m 47s (non-voting)\n- storlets-functional-openjdk17 https://zuul.opendev.org/t/openstack/build/32e336ac1fbf4659804000f91d9b841b : FAILURE in 15m 40s (non-voting)\n- storlets-functional-podman https://zuul.opendev.org/t/openstack/build/15be1ff495fb45ac92b6e5c79f623d99 : FAILURE in 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"136b3c739f9af37238ee2a1de663075a5f5ba09d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 17:38:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"21c10366407f386476200f3ad4f87fd72a9737f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:25:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95ddc179302348cf9676b5d4ac972121\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f788e7d5079c4fb1af3deb3816881b0b : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282d30867eba4fdcaf9ce7b2f8ae8eda : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f0325da358473a84f290b9203f2d40 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e778d46a23b406da3a3a5a947e815d3 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e342466730c4c5da9bc0d59df8ac793 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91260ea757494e0ba571dd33708dd138 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40fd74ac9f1464eaade99db2991045a : SUCCESS in 3m 06s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/337577ac8a974008bcf6468907287604 : SUCCESS in 19m 39s\n- storlets-functional-openjdk21 https://zuul.opendev.org/t/openstack/build/23726804f5414728a04e0f79efb9e6b9 : SUCCESS in 17m 31s (non-voting)\n- storlets-functional-openjdk17 https://zuul.opendev.org/t/openstack/build/d9ca5719ba6a4ed4a72f9fb226b1b10d : SUCCESS in 16m 35s (non-voting)\n- storlets-functional-podman https://zuul.opendev.org/t/openstack/build/3fb564e79d5e4d99b9ca5b98ef6beb2f : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eaded51261d8d3c47bbdced6d25474affd9990a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 18:57:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"994beb0ca48224e0d047750699efa78cff7a06d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:58:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f4d03c7f931432c2800ef5abc04735dd03161e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:14:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c227324c6e4114b90c1e652a36ff72\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ad7d1e826cb458dbf352145ca774b0a : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1dc39ce809b49abaae7045f0c65a2eb : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79aacc176a57475791f564f858e3364e : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44df925c43c5454ba6396d6c09eafa32 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a5b83d9aa3476987a521a4e136fa8d : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d95ecd2ab82a4bd299b2ca7de488a17f : SUCCESS in 5m 07s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/03041d6880204edb912ab6351d103f94 : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a1179779f01de7c1c67e31d67f91a7749afb482e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:14:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"15c170df292ed1be73d7de29dec863b046e4f2ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:15:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902b824b028d43018c24d76a3c958187\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8e4207fe724474b5d27abfe9ab9945 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46f7582473e64e9c8b2396c39ff299df : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87b6e328baf4e6ece469fe28ec5ca188cc60aa85","revisions":{"37c275245f6d069045642851c56d6c1d9ed0e840":{"kind":"REWORK","_number":1,"created":"2026-09-08 17:06:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/1004674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/74/1004674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/74/1004674/1"}}},"commit":{"parents":[{"commit":"0039e11903412d72861553b0c0c08aa5056ba032","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/0039e11903412d72861553b0c0c08aa5056ba032"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:05:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:05:43.000000000","tz":540},"subject":"Replace scripts by module executions","message":"Replace scripts by module executions\n\n... to reduce one mount point.\n\nChange-Id: I5297d2ce4391b40369d4efa1874f9b95f8bda227\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/37c275245f6d069045642851c56d6c1d9ed0e840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/37c275245f6d069045642851c56d6c1d9ed0e840"}]},"branch":"refs/heads/master"},"87b6e328baf4e6ece469fe28ec5ca188cc60aa85":{"kind":"REWORK","_number":2,"created":"2026-09-08 17:38:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/1004674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/74/1004674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/74/1004674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/74/1004674/2"}}},"commit":{"parents":[{"commit":"0039e11903412d72861553b0c0c08aa5056ba032","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/0039e11903412d72861553b0c0c08aa5056ba032"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:05:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:37:45.000000000","tz":540},"subject":"Replace scripts by module executions","message":"Replace scripts by module executions\n\n... to reduce one mount point.\n\nChange-Id: I5297d2ce4391b40369d4efa1874f9b95f8bda227\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/87b6e328baf4e6ece469fe28ec5ca188cc60aa85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/87b6e328baf4e6ece469fe28ec5ca188cc60aa85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
