)]}'
{"id":"openstack%2Fstorlets~1004675","triplet_id":"openstack%2Fstorlets~master~Ic52f0855398b9c6418521a29a6c0a73752902188","project":"openstack/storlets","branch":"master","hashtags":[],"change_id":"Ic52f0855398b9c6418521a29a6c0a73752902188","subject":"Bump java dependencies","status":"MERGED","created":"2026-09-08 17:11:41.000000000","updated":"2026-09-08 17:54:36.000000000","submitted":"2026-09-08 17:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004675","meta_rev_id":"dba5c809970edbf170b376cd8d34586b2280e983","_number":1004675,"virtual_id_number":1004675,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 17:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-08 17:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-08 17:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 17:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d6c2b9154b653040b2404599139e15f8702189e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 17:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1d7c8681f634b31356a6c020d9e7d158b9f3430","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 17:36:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fade8b4ade0b5e0ef2ebb0f65ef5d1550824ad83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7afed5fd59b7469dbdbab709175f7170\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f671345e68498989990c342dc2b03b : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a72d23fdd6814fe98aab4564c5c6fc4d : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/541055d8319c45e5a334259d9a8f55bf : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52c46a8cd1894183952d69381afa5dca : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd148c5e352942be94db02e18f6e6fe9 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2c758da22ff4c6ca70542447dd6a554 : SUCCESS in 2m 55s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/3c85e336e972431aa21eda1d5a629dae : SUCCESS in 17m 32s\n- storlets-functional-openjdk21 https://zuul.opendev.org/t/openstack/build/ec47ae8e660b4132a9e3a8a0ee6606d6 : SUCCESS in 17m 42s (non-voting)\n- storlets-functional-openjdk17 https://zuul.opendev.org/t/openstack/build/c37891f10a0b47348dea1a9feeb43163 : SUCCESS in 20m 25s (non-voting)\n- storlets-functional-podman https://zuul.opendev.org/t/openstack/build/76f5506d1d5b40f5bea2323b7f647207 : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edd1da772ad11d0bc2bf11dec5b93e35c7d51fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3255ce604158d3eb117d1ab0b846b5a94eee8381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:53:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f47a004447b048d39f79e707b1fe469c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f90af01abd42d7af7d3943c3f5152f : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e85c828ea8bc45eab00853aed638036e : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36ea1d5cb626493f9a5f31152faf923f : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008e9ce1358b48e087fe71f9489d4675 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c970ea9c834168a8084f6a19167c73 : SUCCESS in 4m 00s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/09b75ecd45494bc48f2c27084f0289b0 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2446bb3dc49998a59f310baf057c7913b84708b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:53:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dba5c809970edbf170b376cd8d34586b2280e983","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:54:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36c2b1ab788948c299cfe42ac8e17240\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4589281dddd4c05b61af318ab720044 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"584997e338ffd642d0e67b052e1f12551be29276","revisions":{"584997e338ffd642d0e67b052e1f12551be29276":{"kind":"REWORK","_number":1,"created":"2026-09-08 17:11:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/1004675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/75/1004675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/75/1004675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/75/1004675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/75/1004675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/75/1004675/1"}}},"commit":{"parents":[{"commit":"b119b14eef1688dc7a27b8fe0b6f7b06bd9e46f5","subject":"Replace json-simple by gson","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/b119b14eef1688dc7a27b8fe0b6f7b06bd9e46f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:11:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-08 17:11:27.000000000","tz":540},"subject":"Bump java dependencies","message":"Bump java dependencies\n\nChange-Id: Ic52f0855398b9c6418521a29a6c0a73752902188\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/584997e338ffd642d0e67b052e1f12551be29276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/584997e338ffd642d0e67b052e1f12551be29276"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
