)]}'
{"id":"openstack%2Fstorlets~345929","triplet_id":"openstack%2Fstorlets~master~I85953d245422d88e00f7d03c6d02c05ce957a188","project":"openstack/storlets","branch":"master","topic":"refactor_dfactory","hashtags":[],"change_id":"I85953d245422d88e00f7d03c6d02c05ce957a188","subject":"Refactor some process launching functions in daemon factory","status":"MERGED","created":"2016-07-22 09:16:53.000000000","updated":"2016-07-31 06:01:54.000000000","submitted":"2016-07-31 06:01:54.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a5b84d55a40f4fbf1629e2e7ba410e3ec0ea445","_number":345929,"virtual_id_number":345929,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-07-31 06:01:54.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-07-30 19:13:58.000000000","_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-30 19:13:58.000000000","_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-30 19:13:58.000000000","updated_by":{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"},"reviewer":{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"},"state":"REVIEWER"},{"updated":"2016-07-31 06:01:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b8049030aab9307a780491772f00f1788a8c788a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2016-07-22 09:16:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c44384235f1f4d9b14307913ea385e9ca193cfa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2016-07-22 09:26:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7f2982fd0a04e38dd3cee24e1c3ff7e74c45cb1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-22 09:47:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-storlets-pep8 http://logs.openstack.org/29/345929/2/check/gate-storlets-pep8/b774e17/ : SUCCESS in 3m 25s\n- gate-storlets-python27 http://logs.openstack.org/29/345929/2/check/gate-storlets-python27/7f5d017/ : SUCCESS in 2m 48s\n- gate-storlets-functional http://logs.openstack.org/29/345929/2/check/gate-storlets-functional/420e626/ : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a0136aa2ea84c213b12b6324d1ead73e596ea7bb","author":{"_account_id":11317,"name":"Eran Rom","email":"eran@itsonlyme.name","username":"eranr"},"date":"2016-07-30 19:13:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1274842a4f6cb03033d472dfcacfd2ce7dd9b775","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-30 19:14:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"d2a1471e7e4d52e6ab5e0cd3d5a6910e1b296af4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-30 19:37:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-storlets-pep8-ubuntu-xenial http://logs.openstack.org/29/345929/2/gate/gate-storlets-pep8-ubuntu-xenial/4c65f55/ : SUCCESS in 3m 32s\n- gate-storlets-python27-ubuntu-xenial http://logs.openstack.org/29/345929/2/gate/gate-storlets-python27-ubuntu-xenial/a083f6e/ : SUCCESS in 3m 27s\n- gate-storlets-functional http://logs.openstack.org/29/345929/2/gate/gate-storlets-functional/b373b0b/ : FAILURE in 23m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f44e1d4075570920cf36fcdb38bd8d19f2c2e2c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2016-07-31 05:21:46.000000000","message":"Patch Set 2:\n\nreverify","accounts_in_message":[],"_revision_number":2},{"id":"47b87b446f2b574dbaa15fa170e32385a8922b1f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-31 05:41:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-storlets-pep8-ubuntu-xenial http://logs.openstack.org/29/345929/2/check/gate-storlets-pep8-ubuntu-xenial/201af5d/ : SUCCESS in 5m 59s\n- gate-storlets-python27-ubuntu-xenial http://logs.openstack.org/29/345929/2/check/gate-storlets-python27-ubuntu-xenial/31d11f4/ : SUCCESS in 4m 00s\n- gate-storlets-functional http://logs.openstack.org/29/345929/2/check/gate-storlets-functional/074d528/ : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a666b52aba7c17fd8499dca16b6fcb7c0d40c54e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-31 05:41:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"119a174b3e1d20e952b64c6de368a543f397bd87","date":"2016-07-31 06:01:54.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"cb6a0df951c8402fea2e3572a422d4ba16602000","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-31 06:01:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-storlets-pep8-ubuntu-xenial http://logs.openstack.org/29/345929/2/gate/gate-storlets-pep8-ubuntu-xenial/db65d3a/ : SUCCESS in 3m 37s\n- gate-storlets-python27-ubuntu-xenial http://logs.openstack.org/29/345929/2/gate/gate-storlets-python27-ubuntu-xenial/59e4df6/ : SUCCESS in 3m 26s\n- gate-storlets-functional http://logs.openstack.org/29/345929/2/gate/gate-storlets-functional/8afd718/ : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ebd49cf3141a082e9c2a9f9979fa65fb57dfb40","revisions":{"02326df1ccecda545880f3fe17922d8117c8d4b6":{"kind":"REWORK","_number":1,"created":"2016-07-22 09:16:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/345929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/29/345929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/29/345929/1"}}},"commit":{"parents":[{"commit":"12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9","subject":"Refactor unit tests about handler classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@nttdata.co.jp","date":"2016-07-22 09:15:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@nttdata.co.jp","date":"2016-07-22 09:15:41.000000000","tz":540},"subject":"Refactor some process launching functions in daemon factory","message":"Refactor some process launching functions in daemon factory\n\nChange-Id: I85953d245422d88e00f7d03c6d02c05ce957a188\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/02326df1ccecda545880f3fe17922d8117c8d4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/02326df1ccecda545880f3fe17922d8117c8d4b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9","is_merged_in_target_branch":true,"change_id":"I78d4ce2657ec8535acc4f07ba694653c2f53d54f","change_number":342273,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7ebd49cf3141a082e9c2a9f9979fa65fb57dfb40":{"kind":"REWORK","_number":2,"created":"2016-07-22 09:26:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/345929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/29/345929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/29/345929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/29/345929/2"}}},"commit":{"parents":[{"commit":"12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9","subject":"Refactor unit tests about handler classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@nttdata.co.jp","date":"2016-07-22 09:15:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@nttdata.co.jp","date":"2016-07-22 09:25:17.000000000","tz":540},"subject":"Refactor some process launching functions in daemon factory","message":"Refactor some process launching functions in daemon factory\n\nChange-Id: I85953d245422d88e00f7d03c6d02c05ce957a188\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/7ebd49cf3141a082e9c2a9f9979fa65fb57dfb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/7ebd49cf3141a082e9c2a9f9979fa65fb57dfb40"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12f45d8069e7fc80c4eff8dadfbfcc590e62ffa9","is_merged_in_target_branch":true,"change_id":"I78d4ce2657ec8535acc4f07ba694653c2f53d54f","change_number":342273,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
