)]}'
{"id":"openstack%2Fstorlets~542587","triplet_id":"openstack%2Fstorlets~master~Ie25267f07e364d2ba6c18bacaea274c487b7b3dc","project":"openstack/storlets","branch":"master","topic":"baistorlets","hashtags":[],"change_id":"Ie25267f07e364d2ba6c18bacaea274c487b7b3dc","subject":"Fix typos in storlets","status":"MERGED","created":"2018-02-09 04:45:35.000000000","updated":"2018-04-12 10:23:35.000000000","submitted":"2018-04-12 10:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"542587-1523528615561-40add3dc","meta_rev_id":"f2853300c135b38dd52552d628ff20b34cd43f58","_number":542587,"virtual_id_number":542587,"owner":{"_account_id":27678,"name":"baiwenteng","email":"baiwenteng@inspur.com","username":"baiwenteng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2018-04-12 10:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":2,"date":"2018-04-12 09:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":1,"date":"2018-04-12 09:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-12 09:17:57.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2018-04-12 10:23:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b3c6adfb9e4196e2b61d1f820c9d0a71289b121","author":{"_account_id":27678,"name":"baiwenteng","email":"baiwenteng@inspur.com","username":"baiwenteng"},"date":"2018-02-09 04:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d442fe9e3ff66366b5dbc20f043bc931fba5269","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-09 05:10:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/542587/1/check/openstack-tox-pep8/5756c4b/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/87/542587/1/check/openstack-tox-py27/619880c/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/87/542587/1/check/openstack-tox-py35/a27b3c6/ : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/542587/1/check/build-openstack-sphinx-docs/996cdf1/html/ : SUCCESS in 3m 30s\n- storlets-functional http://logs.openstack.org/87/542587/1/check/storlets-functional/4b07c03/ : SUCCESS in 24m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7b5f1f154a84f85f82ad0558b9b8d181bc957a6d","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2018-04-12 09:17:57.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa432058f75c6c926c1ee00e6f6b87e984e09630","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 09:18:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17280e7c749c6faf234b0c354fb8c1e29169fcfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 10:23:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b382aee2f35ed23285e071e0a860f507561195f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 10:23:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/542587/1/gate/openstack-tox-pep8/de06b2b/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/87/542587/1/gate/openstack-tox-py27/ccdacae/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/87/542587/1/gate/openstack-tox-py35/fb5b6b1/ : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/542587/1/gate/build-openstack-sphinx-docs/b1a5bbe/html/ : SUCCESS in 3m 17s\n- storlets-functional http://logs.openstack.org/87/542587/1/gate/storlets-functional/4768231/ : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41903cffe3b338f1daa7e7ced7f8860d506a9312","revisions":{"41903cffe3b338f1daa7e7ced7f8860d506a9312":{"kind":"REWORK","_number":1,"created":"2018-02-09 04:45:35.000000000","uploader":{"_account_id":27678,"name":"baiwenteng","email":"baiwenteng@inspur.com","username":"baiwenteng"},"ref":"refs/changes/87/542587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/87/542587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/87/542587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/87/542587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/87/542587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/87/542587/1"}}},"commit":{"parents":[{"commit":"633c7b9e4894bc075fbc27ce17fe63540a622133","subject":"CHANGELOG and AUTHORS updates for storlets 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/633c7b9e4894bc075fbc27ce17fe63540a622133"}]}],"author":{"name":"baiwenteng","email":"baiwenteng@inspur.com","date":"2018-02-09 04:44:28.000000000","tz":480},"committer":{"name":"baiwenteng","email":"baiwenteng@inspur.com","date":"2018-02-09 04:45:30.000000000","tz":480},"subject":"Fix typos in storlets","message":"Fix typos in storlets\n\nChange-Id: Ie25267f07e364d2ba6c18bacaea274c487b7b3dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/41903cffe3b338f1daa7e7ced7f8860d506a9312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/41903cffe3b338f1daa7e7ced7f8860d506a9312"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"633c7b9e4894bc075fbc27ce17fe63540a622133","is_merged_in_target_branch":true,"change_id":"I5a1a66736e50bb92223c0d3287c44892d45e73fa","change_number":540800,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
