)]}'
{"id":"openstack%2Fstorlets~680398","triplet_id":"openstack%2Fstorlets~master~If671a4f73d8a0279f8ef18938380e3d67ab23add","project":"openstack/storlets","branch":"master","topic":"py3","hashtags":[],"change_id":"If671a4f73d8a0279f8ef18938380e3d67ab23add","subject":"Use binary I/O mode instead of str/unicode","status":"ABANDONED","created":"2019-09-05 12:45:58.000000000","updated":"2019-11-18 05:21:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"caf46948d04c421818241dd3ce95e2abcb25d8f5","_number":680398,"virtual_id_number":680398,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-14 09:06:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-09-14 12:03:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 09:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-14 12:03:00.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"}],"messages":[{"id":"1bce55aae17aac9af235d04984ee195004edab24","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-05 12:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa066406143011ae9732e9fab5283367d49b21ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-05 13:21:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19892c2bb5deed2b142f16bb3cca420d25cd9719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 14:11:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ae22a1e12d40a4a049ea23338c18a1 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc529cdc3fa844d5b782f5757a8fdde3 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca313645a6564c29b8c43f2b51c60f95 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/025859f970e843beb311e6c5e23dda11 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f49a68b22624308b87d57f123c81458 : SUCCESS in 5m 24s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/993ea070bab64336849b68df6a00f937 : SUCCESS in 42m 31s\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/7514f5479d694629bf638388984bc894 : FAILURE in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a45272e1dd6316385a0b6caf5ae44993f27ded85","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-08 02:13:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d4fed6347a3f73d3ca166c630c176c3831625299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 02:54:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a6813c723046a09e97a6517bca76b8 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/726e9d0b919e442da1de574bddde1314 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cef43f69bb14d179b7f24c9b1217f59 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/290d2d5cbc514225932ef0a9456e136f : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65555c2e661443d1830f441c1087bc9e : SUCCESS in 4m 23s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/9129124731214ce78ea3e6cec06147e0 : SUCCESS in 35m 08s\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/02efcbb0088e4449ae8606f1fec51f13 : FAILURE in 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"234ffb30cd65369516b9ef3a081b7cd0d8eb1bf5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-14 08:10:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b80b9773f30b8e97375bb0f6e0efd7886fcde49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 09:06:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53813c1586124292b0d3399dd73adac1 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/308bf486efeb4cada77ca75722db3268 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb25312c356442beaf34c25462662c02 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a52f2a17743d41a58f8aceac470d2724 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01661d077fc84c24bdfd0d08280021b2 : SUCCESS in 5m 13s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/468dd039f6474793b5b805ba70dbd8e8 : SUCCESS in 37m 07s\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/63581a3a7ef940029e792a3ef9b564f3 : FAILURE in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c5d14fdfafa0d8478d10830ef12e5f9b3377319","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:03:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9cded5dd9c75b287386b54be93a1d06b93beb8ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 05:21:48.000000000","message":"Abandoned\n\nAlready done in another patch.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6a7f57d90cbcba20efa66cab816cdea3a8fd4d67","revisions":{"2a4a6d758ce2e8d6405f01adcaa72efd7191e06a":{"kind":"REWORK","_number":1,"created":"2019-09-05 12:45:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/680398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/98/680398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/98/680398/1"}}},"commit":{"parents":[{"commit":"8188824218ab9c9fb346d4c2d5140f15d5ef75dc","subject":"Fix build warning in c codes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8188824218ab9c9fb346d4c2d5140f15d5ef75dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 12:45:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 12:45:13.000000000","tz":540},"subject":"Py3: Pass binary file for put_object","message":"Py3: Pass binary file for put_object\n\nChange-Id: If671a4f73d8a0279f8ef18938380e3d67ab23add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/2a4a6d758ce2e8d6405f01adcaa72efd7191e06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/2a4a6d758ce2e8d6405f01adcaa72efd7191e06a"}]},"branch":"refs/heads/master"},"0e7a63fc5cd29c292c5f983a0a606d82016ad7e0":{"kind":"REWORK","_number":2,"created":"2019-09-05 13:21:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/680398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/98/680398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/98/680398/2"}}},"commit":{"parents":[{"commit":"8188824218ab9c9fb346d4c2d5140f15d5ef75dc","subject":"Fix build warning in c codes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8188824218ab9c9fb346d4c2d5140f15d5ef75dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 12:45:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 13:21:48.000000000","tz":540},"subject":"Py3: Pass binary file for put_object","message":"Py3: Pass binary file for put_object\n\nChange-Id: If671a4f73d8a0279f8ef18938380e3d67ab23add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/0e7a63fc5cd29c292c5f983a0a606d82016ad7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/0e7a63fc5cd29c292c5f983a0a606d82016ad7e0"}]},"branch":"refs/heads/master"},"edb59fccc837e2f8a3a26b0a14a3f89e76f1f85d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-08 02:13:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/680398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/98/680398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/98/680398/3"}}},"commit":{"parents":[{"commit":"c27c1f894b8281daeede9066b768f770de4c86a8","subject":"Fix inappropriate arg passed to nostest command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c27c1f894b8281daeede9066b768f770de4c86a8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 12:45:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-08 02:13:26.000000000","tz":540},"subject":"Py3: Pass binary file for put_object","message":"Py3: Pass binary file for put_object\n\nChange-Id: If671a4f73d8a0279f8ef18938380e3d67ab23add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/edb59fccc837e2f8a3a26b0a14a3f89e76f1f85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/edb59fccc837e2f8a3a26b0a14a3f89e76f1f85d"}]},"branch":"refs/heads/master"},"6a7f57d90cbcba20efa66cab816cdea3a8fd4d67":{"kind":"REWORK","_number":4,"created":"2019-09-14 08:10:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/680398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/98/680398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/98/680398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/98/680398/4"}}},"commit":{"parents":[{"commit":"c27c1f894b8281daeede9066b768f770de4c86a8","subject":"Fix inappropriate arg passed to nostest command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c27c1f894b8281daeede9066b768f770de4c86a8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-05 12:45:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-14 08:09:43.000000000","tz":540},"subject":"Use binary I/O mode instead of str/unicode","message":"Use binary I/O mode instead of str/unicode\n\nChange-Id: If671a4f73d8a0279f8ef18938380e3d67ab23add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/6a7f57d90cbcba20efa66cab816cdea3a8fd4d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/6a7f57d90cbcba20efa66cab816cdea3a8fd4d67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
