)]}'
{"id":"openstack%2Fstorlets~698226","triplet_id":"openstack%2Fstorlets~master~I07baf5f0fb553c3c959ea67bcd91f584ae110f14","project":"openstack/storlets","branch":"master","topic":"py3-agent","hashtags":[],"change_id":"I07baf5f0fb553c3c959ea67bcd91f584ae110f14","subject":"DNM: Just testing","status":"ABANDONED","created":"2019-12-10 12:25:47.000000000","updated":"2019-12-18 07:00:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08bcd24648e8986cff6cd4c8c8af05c86e12036c","_number":698226,"virtual_id_number":698226,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-10 13:50:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2019-12-10 12:25:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 13:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc9ed518d9c16ece585446f1c8631f06734f0b57","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-12-10 12:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d3dde5131cb91d8b4b2f30549795a54c2110de5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-12-10 12:25:58.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"993762e65e12d8284513d7a645e02384acca83e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:00:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93ee3e780ed4b5294445ea3570fa63b : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a586aa625e9473f87c611d199e0f75e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd6dadac89e84bc8a1cc62b7eec9b93a : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/388bad6444f24f87b416b75e6fbd20bf : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624143a830c4468e826fe5736f252720 : SUCCESS in 5m 50s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/80e80276a4ab4363aee90c32e2ba205c : TIMED_OUT in 32m 22s\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/91f840d76a9b43dfb16c8d94b626c78a : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"08e99d95ce67a828db1620310f191b5a6a9ff385","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-12-10 13:22:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82dfeb9ba6756b4bf521c270f24ec81e4b2d899f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:50:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e71d8c32cf440c28c4914170cab1c99 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de172194193544a782e1bf15bf527c6c : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cac84b0b39c04f299a374172599dbf2c : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11d1b04897a843b5995084305f689eb1 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20897b502494189891267aca45f7698 : SUCCESS in 6m 42s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/a27f0af9c883407382e5404823cb83f1 : SUCCESS in 20m 40s\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/4f41196f61ce4dbbbd34c7f438601749 : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3feccc2c1aacca9ce69d034e9ad4a50527ca125f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-12-18 07:00:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae60d2ac6f46b2f3d2f3d87cd9c6569724c7fa5c","revisions":{"ae60d2ac6f46b2f3d2f3d87cd9c6569724c7fa5c":{"kind":"REWORK","_number":1,"created":"2019-12-10 12:25:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/698226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/26/698226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/698226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/698226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/698226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/26/698226/1"}}},"commit":{"parents":[{"commit":"ade5040b82a006add13d9384ad69b709df1a44f1","subject":"Merge \"Ensure task_id is a native string on py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ade5040b82a006add13d9384ad69b709df1a44f1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-13 00:49:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-12-10 12:25:32.000000000","tz":540},"subject":"DNM: Just testing","message":"DNM: Just testing\n\nChange-Id: I07baf5f0fb553c3c959ea67bcd91f584ae110f14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ae60d2ac6f46b2f3d2f3d87cd9c6569724c7fa5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ae60d2ac6f46b2f3d2f3d87cd9c6569724c7fa5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
