)]}'
{"id":"openstack%2Fstorlets~744386","triplet_id":"openstack%2Fstorlets~master~Ic75b1672a4689225a92e67064cc64f9183cc4c0d","project":"openstack/storlets","branch":"master","topic":"focal","hashtags":[],"change_id":"Ic75b1672a4689225a92e67064cc64f9183cc4c0d","subject":"Use Ubuntu Forcal as base container image","status":"ABANDONED","created":"2020-08-03 00:46:16.000000000","updated":"2021-06-26 17:52:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e295545dac1bc8b06be3ebbdefee731e0334cd2f","_number":744386,"virtual_id_number":744386,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-25 01:16:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 01:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52894c94eb93b7824b82b39ca694c954a26d9d34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-03 00:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02110f495ab9c3a2dcda722848746f0cba617b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 01:51:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a380aa19a98d4e1686522275f76fe1df : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b55beea79add4ed3abcb9c386bdd7019 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d95006d5edd41e1aae36758046714a5 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f811d2987f944662892005c58c0572a6 : FAILURE in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6b57bc17774533b4e05c177ca934b4 : SUCCESS in 4m 48s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/589d90ca23c144e0b7f65580a13200a5 : RETRY_LIMIT in 11m 32s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/82610c53983e428eb0841deaa2f01b95 : RETRY_LIMIT in 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"67b08cfca4e320d367dd83266292659fc11a02e6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-20 15:11:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c4902240e3a4cedf05a390d51f6dcc17f595bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:47:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ca509cc50a46f6a8329c7fab7540a4 : RETRY_LIMIT in 1m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0a30d72d7f24670be1eda927e94c079 : RETRY_LIMIT in 1m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3f5634e5254cd2befc7a88bae92b40 : RETRY_LIMIT in 1m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5227393eb54d85a95ed7211c171c5f : RETRY_LIMIT in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e8c839d0dc4ad283047c19cbf81018 : RETRY_LIMIT in 1m 48s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/aabac61f9c764a069818f07b4a4cdf88 : RETRY_LIMIT in 1m 44s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/d17de573d4ae413d9e244148b7045659 : RETRY_LIMIT in 1m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1c6fa6275ca2e8531e2f7106883f94404c4e58c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-24 14:32:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cd12a45c766948f75e3b21c618f9cac8a5002fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:30:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ddc99480eb4e59b3469647ef4c11dc : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79d57743292c4deeadb949fc70874181 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d2cbf622ba46a1aa18bc8eb3e24101 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836935b2f4154763a736a74ba2a4293b : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8421fa906449d3adfd97a3bc134270 : SUCCESS in 5m 57s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/621733394b92436baff826853b2db3c0 : RETRY_LIMIT in 13m 00s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/36e32baaf37d40658832df02431e6620 : RETRY_LIMIT in 12m 13s","accounts_in_message":[],"_revision_number":2},{"id":"85e4f3388bdd924d6df3e24ce93d97438a0a6078","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-25 00:37:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9712dcbf745cdfcf2beca2874549b6e65b9a6bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 01:16:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e9b81a700d4479a4d9c431ade61db0 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e68c78798a1542ec90906f7321ccdf33 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9361d46bc389403798bb01283d5db9b0 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e97491e82cc54840a6d0303847a0c7ff : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf6896d5496483abce871d2017195a3 : SUCCESS in 5m 06s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/e09f5ad4ed9842b6b6aa3c6f66a2aa21 : FAILURE in 26m 33s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/e170bb0256f54d1bb92cd0ffd3c60817 : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e295545dac1bc8b06be3ebbdefee731e0334cd2f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-26 17:52:46.000000000","message":"Abandoned\n\nThis was done by https://review.opendev.org/c/openstack/storlets/+/761155","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a67ded5ad44cb766a5f5b60bbccbdd023f727aa3","revisions":{"3bbf903232f401fb6775e0380a289bd1ad9fcd2f":{"kind":"REWORK","_number":1,"created":"2020-08-03 00:46:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/744386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/86/744386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/86/744386/1"}}},"commit":{"parents":[{"commit":"8462e4e688af70cf900ed72454092f3b4a11b13f","subject":"Merge \"Small cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8462e4e688af70cf900ed72454092f3b4a11b13f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-03 00:44:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-03 00:44:39.000000000","tz":540},"subject":"Use Ubuntu Forcal as base container image","message":"Use Ubuntu Forcal as base container image\n\nChange-Id: Ic75b1672a4689225a92e67064cc64f9183cc4c0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/3bbf903232f401fb6775e0380a289bd1ad9fcd2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/3bbf903232f401fb6775e0380a289bd1ad9fcd2f"}]},"branch":"refs/heads/master"},"c05156c44ca5371bee4c49dcff24553f233c82dd":{"kind":"REWORK","_number":2,"created":"2020-08-20 15:11:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/744386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/86/744386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/86/744386/2"}}},"commit":{"parents":[{"commit":"8462e4e688af70cf900ed72454092f3b4a11b13f","subject":"Merge \"Small cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8462e4e688af70cf900ed72454092f3b4a11b13f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-03 00:44:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-20 15:10:50.000000000","tz":540},"subject":"Use Ubuntu Forcal as base container image","message":"Use Ubuntu Forcal as base container image\n\nChange-Id: Ic75b1672a4689225a92e67064cc64f9183cc4c0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c05156c44ca5371bee4c49dcff24553f233c82dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c05156c44ca5371bee4c49dcff24553f233c82dd"}]},"branch":"refs/heads/master"},"a67ded5ad44cb766a5f5b60bbccbdd023f727aa3":{"kind":"REWORK","_number":3,"created":"2020-08-25 00:37:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/744386/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/86/744386/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/86/744386/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/86/744386/3"}}},"commit":{"parents":[{"commit":"8462e4e688af70cf900ed72454092f3b4a11b13f","subject":"Merge \"Small cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8462e4e688af70cf900ed72454092f3b4a11b13f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-03 00:44:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-25 00:36:55.000000000","tz":540},"subject":"Use Ubuntu Forcal as base container image","message":"Use Ubuntu Forcal as base container image\n\nChange-Id: Ic75b1672a4689225a92e67064cc64f9183cc4c0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a67ded5ad44cb766a5f5b60bbccbdd023f727aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a67ded5ad44cb766a5f5b60bbccbdd023f727aa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
