)]}'
{"id":"openstack%2Fstorlets~750223","triplet_id":"openstack%2Fstorlets~master~I86bb88fcd556326d2fa70f9a7a7c6f770c224361","project":"openstack/storlets","branch":"master","topic":"docker-nosudo","hashtags":[],"change_id":"I86bb88fcd556326d2fa70f9a7a7c6f770c224361","subject":"Use Docker Python SDK to manage containers","status":"ABANDONED","created":"2020-09-08 00:09:30.000000000","updated":"2020-09-08 10:16:45.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98fd497a9123e87f5b280458072bfaa7dedd2cc3","_number":750223,"virtual_id_number":750223,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-08 09:51:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 09:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c7f06419f13123fa08df3f5effcd7af7076e2e6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 00:09:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf0c7803cfd0944f80e05444c3a30016f23e5fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 00:42:23.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05886b194a4141a481a5a6b5750c011d : FAILURE in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffe42ce150ec42ed89791f566f655d1c : FAILURE in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6062bbb8434aadb1316d5cff396bb9 : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/856bd999a05f4ec4aee5eff0f6b97418 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861b63b6e7e64b9eb69ce6d6c5cb178b : SUCCESS in 5m 25s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/0359f0b774314b279af38e67032f1668 : RETRY_LIMIT in 1m 36s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/fb24c3da6871483a97944a33748f055f : RETRY_LIMIT in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2688741675235d47d20698dcf660fc4a382a0e5a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 03:32:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aba943c7b1fd39e59e834be0f0241a87da49480b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 04:18:19.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd636537251f459caf1e1e1cc72650cf : FAILURE in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e46e835a1dc4ac6b1424ab02e88520a : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7997024a2bb3453ca949f81a00e5f7ca : FAILURE in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ecd5cb024d48dcb4b70d9edff0abde : FAILURE in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ccea9a17a94eadb691f2b18fb9af2a : SUCCESS in 5m 10s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/9c1cef63036d4d53bf8b4c475200f252 : RETRY_LIMIT in 13m 48s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/a858f88883fc48cf8a70cfc10f4fabff : RETRY_LIMIT in 13m 08s","accounts_in_message":[],"_revision_number":2},{"id":"dad7289dccee50d84bb2bfad68f1feed41fc12b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 04:28:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"006b827a478976442697fe8d7d4f5d056a00d452","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 04:39:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e7b6346a6669fbd2b1aa866a4784fd636f5a26bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 05:00:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407671d2d3b54d9cb27e35b8982db761 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4b8ff30e23e4ea69e29b91990fdcbe4 : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d30b998576c46acb587610fa0f0da34 : FAILURE in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bad7f5f41624510b7395851e7ca64a2 : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b88cc944c54b1dad2598054134e1b0 : SUCCESS in 5m 10s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/c146c4ed9c8b49deae1e2a99de17867e : FAILURE in 15m 52s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/6823545753f745c0bd8e18dd532b567a : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":4},{"id":"833cac3147b97999cc4131d227123a7a56794312","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 05:51:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91ae53bccdbc7880984e6f8c90b9b2dcca193dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 06:12:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f33ea18fa44b78af939344608e4b22 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff9a9524269d488fbad2f57e12d55b23 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9ff4c96b47f4f6dac9365e7d0dce034 : FAILURE in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2880d66d4979448888c10e62d50a918e : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5533da759064718b791bfba39133e6d : SUCCESS in 4m 11s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/d43995fe5a3145e388ea1dc4afcb9fe5 : FAILURE in 20m 19s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/cc167879965a4a99b003e57b36c08bb7 : FAILURE in 20m 01s","accounts_in_message":[],"_revision_number":5},{"id":"d7b8b3cdaffb52d48cf365114c72f90c1e6cdbd9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 06:33:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"db0716b562a617b2aee921fc9bbb4a37794e3a9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 06:47:18.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"9c44d6593f5c00ad2d2c72dc90308ac0d9faa631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 07:14:33.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ce0de983d54e72b7622a43cadbcc6d : FAILURE in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf7757ff0e6b4e1884681122706f5439 : FAILURE in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4538bd8a8d7c4cecbb71f75b243b3be7 : FAILURE in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a71d4c695544e0a65ea851fd2da127 : FAILURE in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32686381c0154b7abb4f77b18beacbdd : SUCCESS in 6m 25s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/5664580ea1a54e4094b66e4c34ab2faa : FAILURE in 14m 26s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/474931b46fcf4197a232d534d38d0245 : FAILURE in 17m 59s","accounts_in_message":[],"_revision_number":7},{"id":"3913dc02d3be0736b8570a594c59fc78be8bf595","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 07:36:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93e79275399b5d5acd173af4e1558f8dffacda1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 07:59:10.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87e1202da544cadad416c75d18e5b65 : FAILURE in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89e0ab0ed77343bbaf5561033779c425 : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfb4ff1363584d4590ea4621b2b56e0f : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f045aa1bf8e7475fa0dbf64d3f6caa14 : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9665464b341847f8b6af02eba973e988 : SUCCESS in 5m 21s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/eb7ca2f14b1541d092d327b95b6d277f : SUCCESS in 21m 07s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/7ed3df034d7a411980ece8361452c0cd : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":8},{"id":"792d07b8022154fba4c96de5661479b9eb7406dd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 08:03:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"da13159cebbfea69788262c3cfd859c9416d22dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 08:21:11.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5cac8fa23447489526a9149e6e6086 : FAILURE in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b7f829d21b84205a53a9a992c54406d : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b6c1552112148e1b5ee386a5cb1f994 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f604a89224c4e53a916f6e3e4682439 : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc5c2f1a20a4a8f8373c0c105c02db1 : SUCCESS in 5m 14s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/e5fa06f4018b48de8dbad8378a7a7f56 : SUCCESS in 16m 26s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/2a5fe611f9c1470c855bbd41a98cc47a : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":9},{"id":"2ca8aa1b4c73ffde2129f4a6bc2c6e3c9a7a41d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 08:29:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"af2096ab20c3d89e11223913d1c022a0c255a6e7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 08:32:07.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bbb1b92bac7cf8450a3bbade5d5077fb2df037e2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 08:54:57.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c415a5810e0bd75a393cfd1c95eebaa155911eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:25:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980f17e7a65d4a8e93fc3c20de89adc8 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3eecb9db6084c5cb60fa4d81658ca43 : FAILURE in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9ef91dff6964f43bca2d072b86f4499 : FAILURE in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/407958d4062e4e11b051e589618ea1cc : FAILURE in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba58ab7ee764c01b228ce2372ee809c : SUCCESS in 5m 45s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/d1bd970cf5504d47911fd8b771a4b1a9 : SUCCESS in 20m 14s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/fd8356204f8a46049abe174b8d72a700 : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":12},{"id":"dbf5c5a25b4ebe2a71047ce2ed3ed38cb4438687","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 09:26:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1df3bf34b9824076523b17c018fd8f2d42864f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:51:07.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de37f9465a404a7fa693e37bfd74932a : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eb499583e344a0097732c323c598ab2 : FAILURE in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c27eb83d61974389940c06f42df0af09 : FAILURE in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d180bf6777d64eefb3e55e64eb8c7b17 : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e14e57e02443b288e6a720ff5e6120 : SUCCESS in 6m 17s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/71ed972b2eb9418cbf183c430df718b9 : SUCCESS in 16m 47s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/4c608098db91425bac8386228b639adc : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":13},{"id":"34c3fc337b0bbef427c885f436f5f5b770c8b6b6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-08 10:16:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"71f1584a5d3a0e9548b1116c8b543be76d26d889","revisions":{"a3d84757eb9d774d048a5874135664966ccc7a2d":{"kind":"REWORK","_number":1,"created":"2020-09-08 00:09:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/1"}}},"commit":{"parents":[{"commit":"8ed94a646ca8dc978e3b1752c9ac0172c69db48b","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/8ed94a646ca8dc978e3b1752c9ac0172c69db48b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:08:48.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a3d84757eb9d774d048a5874135664966ccc7a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a3d84757eb9d774d048a5874135664966ccc7a2d"}]},"branch":"refs/heads/master"},"aba2b32114b993dc67034d0c3ecc4a3bb4c6bb35":{"kind":"REWORK","_number":2,"created":"2020-09-08 03:32:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/2"}}},"commit":{"parents":[{"commit":"4a8439aa58de69285c98dc5bf887ab2f6d581c15","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/4a8439aa58de69285c98dc5bf887ab2f6d581c15"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 03:32:33.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/aba2b32114b993dc67034d0c3ecc4a3bb4c6bb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/aba2b32114b993dc67034d0c3ecc4a3bb4c6bb35"}]},"branch":"refs/heads/master"},"3500240236a3311764c0144514d6b750e6c3fb9b":{"kind":"REWORK","_number":3,"created":"2020-09-08 04:28:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/3"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 04:27:48.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/3500240236a3311764c0144514d6b750e6c3fb9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/3500240236a3311764c0144514d6b750e6c3fb9b"}]},"branch":"refs/heads/master"},"a476e48b8dacc90c0a73c64cdfd57a38783990e9":{"kind":"REWORK","_number":4,"created":"2020-09-08 04:39:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/4"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 04:38:30.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a476e48b8dacc90c0a73c64cdfd57a38783990e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a476e48b8dacc90c0a73c64cdfd57a38783990e9"}]},"branch":"refs/heads/master"},"53d02499838a0aa760275b2490fd392eee95d4e9":{"kind":"REWORK","_number":5,"created":"2020-09-08 05:51:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/5"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 05:23:12.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/53d02499838a0aa760275b2490fd392eee95d4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/53d02499838a0aa760275b2490fd392eee95d4e9"}]},"branch":"refs/heads/master"},"5d243cf8ba46ab515ccb33704f4a88e5e62619d9":{"kind":"REWORK","_number":6,"created":"2020-09-08 06:33:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/6"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 06:33:00.000000000","tz":540},"subject":"WIP: Use Docker Python SDK to manage containers","message":"WIP: Use Docker Python SDK to manage containers\n\nTODO: fix unit tests\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/5d243cf8ba46ab515ccb33704f4a88e5e62619d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/5d243cf8ba46ab515ccb33704f4a88e5e62619d9"}]},"branch":"refs/heads/master"},"e5d0a71746d8f9660d02cc269f8d0f1f52d44f6b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-09-08 06:47:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/7"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 06:46:08.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/e5d0a71746d8f9660d02cc269f8d0f1f52d44f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/e5d0a71746d8f9660d02cc269f8d0f1f52d44f6b"}]},"branch":"refs/heads/master"},"c5aa9e112314a2ea5fe8157b736892b65d981836":{"kind":"REWORK","_number":8,"created":"2020-09-08 07:36:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/8"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 07:35:54.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c5aa9e112314a2ea5fe8157b736892b65d981836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c5aa9e112314a2ea5fe8157b736892b65d981836"}]},"branch":"refs/heads/master"},"361f412c90a950d5930cef96d08ef940bf1e8226":{"kind":"REWORK","_number":9,"created":"2020-09-08 08:03:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/9"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 08:02:49.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/361f412c90a950d5930cef96d08ef940bf1e8226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/361f412c90a950d5930cef96d08ef940bf1e8226"}]},"branch":"refs/heads/master"},"a8f12a1d79c96ffb981190c1f255b13993b16ce0":{"kind":"REWORK","_number":10,"created":"2020-09-08 08:29:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/10"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 08:29:01.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a8f12a1d79c96ffb981190c1f255b13993b16ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/a8f12a1d79c96ffb981190c1f255b13993b16ce0"}]},"branch":"refs/heads/master"},"9665dcd57cfea892ae0af26990fdbac2769a8afa":{"kind":"REWORK","_number":11,"created":"2020-09-08 08:32:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/11"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 08:31:43.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/9665dcd57cfea892ae0af26990fdbac2769a8afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/9665dcd57cfea892ae0af26990fdbac2769a8afa"}]},"branch":"refs/heads/master"},"fdc296c31bedf4c3afdba1046ffcb64702ef1c16":{"kind":"REWORK","_number":12,"created":"2020-09-08 08:54:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/12"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 08:54:38.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/fdc296c31bedf4c3afdba1046ffcb64702ef1c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/fdc296c31bedf4c3afdba1046ffcb64702ef1c16"}]},"branch":"refs/heads/master"},"71f1584a5d3a0e9548b1116c8b543be76d26d889":{"kind":"REWORK","_number":13,"created":"2020-09-08 09:26:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/750223/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/23/750223/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/23/750223/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/23/750223/13"}}},"commit":{"parents":[{"commit":"ee415273fbadbbc032b81f4abf406cbfc8a0a911","subject":"Remove SUID flag from restart_docker_container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/ee415273fbadbbc032b81f4abf406cbfc8a0a911"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 00:04:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-08 09:25:55.000000000","tz":540},"subject":"Use Docker Python SDK to manage containers","message":"Use Docker Python SDK to manage containers\n\nThis patch reimplements docker container management by python, because\nwe no longer require SUID binary built from C.\nThis helps us to implement some more logics based on configuration\nparameters and also add some more test coverage.\n\nChange-Id: I86bb88fcd556326d2fa70f9a7a7c6f770c224361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/71f1584a5d3a0e9548b1116c8b543be76d26d889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/71f1584a5d3a0e9548b1116c8b543be76d26d889"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
