)]}'
{"id":"openstack%2Fstorlets~773226","triplet_id":"openstack%2Fstorlets~master~I2843bf1118505a2e32e1f80e7329ed73de1cd62d","project":"openstack/storlets","branch":"master","topic":"restart-log","hashtags":[],"change_id":"I2843bf1118505a2e32e1f80e7329ed73de1cd62d","subject":"Add warning log when restarting docker container","status":"MERGED","created":"2021-01-30 06:46:32.000000000","updated":"2021-03-25 11:17:06.000000000","submitted":"2021-03-25 11:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773226-restart-log","meta_rev_id":"11438f8f3a4dad93f540ad497001628984a86d43","_number":773226,"virtual_id_number":773226,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 11:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-03-25 10:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-03-25 10:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"}],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-30 07:10:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 10:44:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"}],"messages":[{"id":"011a0ffae2100bc6c5524b07808137b81971b7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-30 06:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79ebbda8317b039a3142675ed3df2ae6bb65cd28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 07:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9dfb96268ef4867a39f8d0526a68747 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd7155baec964ef4bb42fa08c4e0b2c2 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d570ea111b4183822c9abc6b4cd639 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb0eb00703a24af0b3475471461d804e : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aa68160eddd4f469fa333a4d782fe33 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83452414b2af46f1ae4427af40a7dcf5 : SUCCESS in 4m 23s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/19b77a6836c54fa8a153b148046ffa97 : SUCCESS in 18m 52s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/584736f566084346a4f5ed2a5a8c8e42 : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6a3eb87428db20f8415d3d7d0d1f1844a26adbf3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-25 10:44:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis change doesn\u0027t affect behavior but makes some changes in logging for better visibility. So I\u0027d approve this by myself now.","accounts_in_message":[],"_revision_number":1},{"id":"8d1bab890815dda377acced557adcbc9f2b99d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 10:44:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83ca71b3779c768038b5884727e1966edf8e8d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:15:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3024c9c60e01464cb17f3de718b2e275 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c918d390140b4ae6b77dd9b1c1e7cba3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99a723ea29d34d51b3d7095bb4373359 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9dcdfea087f454ca608d9d20bdaa726 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1605788181044628dc3c5cf85ed238e : SUCCESS in 5m 21s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/29e6474f14f0452abc6d8dd21ea77dce : SUCCESS in 22m 25s (non-voting)\n- storlets-functional-py3 https://zuul.opendev.org/t/openstack/build/895df8220aec49a8955ba9dd8b92c370 : SUCCESS in 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"492c585bf420e3834c2b68083dc5ea77ec41e3c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11438f8f3a4dad93f540ad497001628984a86d43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:17:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebc30757b184fe9a32384e7d5eb0b71 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f67a8c42cbccc5919cbb1c307b1169045e48b4d","revisions":{"1f67a8c42cbccc5919cbb1c307b1169045e48b4d":{"kind":"REWORK","_number":1,"created":"2021-01-30 06:46:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/773226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/26/773226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/773226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/773226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/26/773226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/26/773226/1"}}},"commit":{"parents":[{"commit":"93fc96b3b1793522cadcc5b2145dc84e789376b7","subject":"Revert \"remove py37\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/93fc96b3b1793522cadcc5b2145dc84e789376b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-30 06:16:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-30 06:46:25.000000000","tz":540},"subject":"Add warning log when restarting docker container","message":"Add warning log when restarting docker container\n\nWhen a storlet application or its dependencies are updated, we first\ntry restarting storlet daemon, but if it fails then we restart the\nwhole container, which might cause ungraceful shutdown of the other\nstorlet daemon in the same container.\n\nThis change adds a warning log so that operators can understand how\nmany times such stop failure have happened in the deployment.\n\nChange-Id: I2843bf1118505a2e32e1f80e7329ed73de1cd62d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/1f67a8c42cbccc5919cbb1c307b1169045e48b4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/1f67a8c42cbccc5919cbb1c307b1169045e48b4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
