)]}'
{"id":"openstack%2Fstorlets~957242","triplet_id":"openstack%2Fstorlets~master~If93917303ca993dba6762cbe2310b09c2919da1d","project":"openstack/storlets","branch":"master","hashtags":[],"change_id":"If93917303ca993dba6762cbe2310b09c2919da1d","subject":"Do not swallow stdout from storlet daemons","status":"MERGED","created":"2025-08-13 13:53:20.000000000","updated":"2025-08-15 01:22:03.000000000","submitted":"2025-08-15 01:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957242","meta_rev_id":"928eee90ddaf8a6fdb66855f566c7868b3863e49","_number":957242,"virtual_id_number":957242,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 01:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-08-14 11:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-08-14 11:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 16:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a2fbf2f9917092f9c2016441b23c7d3db9ffe33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 13:53:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"016e6cd3f5ccd68429c39449dbfc14010524769c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 15:29:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b0b190087b8ec37d087f60e2a23b4b7f2592d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 16:11:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c2e46affef44889a3a5bdee46dd06e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e8c422c5e849f3b984f430489dfff0 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef6418a1c9f749c8ad1a8e9a666c5607 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4d2c9e04fca44a08416832b9ec86e50 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3e75ef7f6894b329b03ed5b2a3814fd : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159b50302f4b4e3b97aa9af83cac0f8c : SUCCESS in 4m 16s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/7f119c23b1924df3a93dbf533e494d0a : SUCCESS in 17m 39s\n- storlets-functional-podman https://zuul.opendev.org/t/openstack/build/d2859b7ca685416dafc838772b342c0f : SUCCESS in 17m 51s (non-voting)\n- storlets-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/98a7fd52fabc477094f403d258c245d5 : RETRY_LIMIT in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54a5acacc70da2ce1f87916b3ffe93848360f96f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 11:43:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f145985ca09cc7347c9fc9dff75a0cff233e6b17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 11:49:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d563a3b241ef5625c87883b9bb36e8bff3f5638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:21:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90511ade4f824a38946db7d57a74b4fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8256d359b46743f7b699ee3e73e96076 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f2c6b478eea4dafb5943aefc9c9dcfe : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d7904596bf54ff988955e8d23f15707 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823f13ddff9942a18038a3561ef1d1ba : SUCCESS in 3m 35s\n- storlets-functional https://zuul.opendev.org/t/openstack/build/f078ad24f95c4a8a977ceda81e50ca39 : SUCCESS in 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"482568a4042d24436fd8b927780d9da3b075e8e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:21:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"928eee90ddaf8a6fdb66855f566c7868b3863e49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:22:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f69d1d57aa4428c9a7204ad6cc34f1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca59f69accb4451864a8f54c5e07d43 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c911bc32fd60a5dc4828a77bdf06e3f165fcc263","revisions":{"4b4f878b32ddc1f6e33e7d81cee4b0a43e7aa5c2":{"kind":"REWORK","_number":1,"created":"2025-08-13 13:53:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/957242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/42/957242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/42/957242/1"}}},"commit":{"parents":[{"commit":"da318b97e5851220ee988513293a2ce42ee52e44","subject":"Merge \"Refactor container-specific runtime classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/da318b97e5851220ee988513293a2ce42ee52e44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-13 13:52:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-13 13:53:08.000000000","tz":540},"subject":"Do not swallow stdout from storlet daemons","message":"Do not swallow stdout from storlet daemons\n\nChange-Id: If93917303ca993dba6762cbe2310b09c2919da1d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/4b4f878b32ddc1f6e33e7d81cee4b0a43e7aa5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/4b4f878b32ddc1f6e33e7d81cee4b0a43e7aa5c2"}]},"branch":"refs/heads/master"},"c911bc32fd60a5dc4828a77bdf06e3f165fcc263":{"kind":"REWORK","_number":2,"created":"2025-08-13 15:29:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/957242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/storlets","ref":"refs/changes/42/957242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/storlets refs/changes/42/957242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/storlets refs/changes/42/957242/2"}}},"commit":{"parents":[{"commit":"da318b97e5851220ee988513293a2ce42ee52e44","subject":"Merge \"Refactor container-specific runtime classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/da318b97e5851220ee988513293a2ce42ee52e44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-13 13:52:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-13 15:29:26.000000000","tz":540},"subject":"Do not swallow stdout from storlet daemons","message":"Do not swallow stdout from storlet daemons\n\nChange-Id: If93917303ca993dba6762cbe2310b09c2919da1d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c911bc32fd60a5dc4828a77bdf06e3f165fcc263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/storlets/commit/c911bc32fd60a5dc4828a77bdf06e3f165fcc263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
