)]}'
{"id":"openstack%2Fsushy-cli~712447","triplet_id":"openstack%2Fsushy-cli~master~Ic7737b92c70af738c7998e412cb59d149418542e","project":"openstack/sushy-cli","branch":"master","topic":"10-add-prefix-option","hashtags":[],"change_id":"Ic7737b92c70af738c7998e412cb59d149418542e","subject":"Add non-default Redfish service mount location","status":"MERGED","created":"2020-03-11 14:12:29.000000000","updated":"2020-03-24 11:08:45.000000000","submitted":"2020-03-24 08:51:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712447-1585039903060-d1a6a7d1","meta_rev_id":"6d21baad0f565cb79de024297d7c68a5b7c49cad","_number":712447,"virtual_id_number":712447,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 08:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-03-12 07:50:08.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-24 08:41:06.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-24 08:41:06.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 07:50:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-24 08:41:06.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-03-24 08:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31948069d8e3cc4ede0dea7d0bd9080148d05eb2","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-11 14:12:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7759064a4a1f169e2d6d9a8485402fec415c57b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 14:46:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e99d9aec5d904a539927c4757567eb0f : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/859410363aa7417a9c4528beddf36ce7 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2caa983d2a042cdb6e680a9cfc36f2c : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ded96833f3427aa87cbbb93d01d14b : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff59adc096d840c888e79bf271603495 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5831265594ad47f69c65cc492eef7a95 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9332fef8894729add13762f8726b06 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"89d15b586f064008cfee4cbddb24b437c251cf1a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-11 17:05:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"253157ced030135df28763db64e7268747ba88ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:20:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b273ebb8a8064a4ca150d129789b25b7 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c32b4dea5654b919516e448cdb2d833 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca7e164963b49f492e401662a91421d : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852972bf695048d9908590823cd68c79 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a79b511fc8454e568d83fbd1ea8f054c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74af8d9105fb4f11930d0c3e273518cd : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553b1bdaee4a4a72926ca50d70aa098f : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6fd7b9d1a695f1c0934742b8c182e8d669d7b18e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-12 07:50:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"10b2fa852686d32009387571a0a28c2a2664ee89","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-24 08:41:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf7b71d6a88cc1e3364f5a8e5d7845c4d0c42dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:41:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ca65092265950d31d0a62266ad58515e3cf3f81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:51:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3cc7246c33f49c98c1ab06573d55949 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08885f7cfea8473ab8ea7b36b07329e0 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a6eceae9bd348c0a804683a018a80bf : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af5cf429136b4a728d3dac74243904ac : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf84bb11a91402896a79bff9d2bb974 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4430906abd0d25365b1a2ebaedf66aff0dd4e673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:51:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2480dac85dd21dd5ad5ff5520fb06b7bd64e9a43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:52:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5354321e88c4c3297d28a122ed38ca9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9000023c35087d3bbd4e22235dbc42b2e97448be","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-24 09:03:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"650d26a8867fbdc6da3e3ff3024c07c7bb41ac61","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-24 11:08:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"411faf1d857ffbe5600c0632ac86401b62806bf5","revisions":{"4281a37166d70ad184eb07ca6cc2428616151ba7":{"kind":"REWORK","_number":1,"created":"2020-03-11 14:12:29.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/47/712447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/47/712447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/1"}}},"commit":{"parents":[{"commit":"7d18cb1fd73e072259c21b86d5ed7a05d28d13b8","subject":"Add `--insecure` and `--tls-certificates` options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/7d18cb1fd73e072259c21b86d5ed7a05d28d13b8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-11 13:57:56.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-11 14:12:09.000000000","tz":60},"subject":"Add non-default Redfish service mount location","message":"Add non-default Redfish service mount location\n\nThe `--service-endpoint` option can now contain a path\nto Redfish API mount point. If not set, it defaults\nto sushy default which is `redfish/v1`.\n\nChange-Id: Ic7737b92c70af738c7998e412cb59d149418542e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/4281a37166d70ad184eb07ca6cc2428616151ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/4281a37166d70ad184eb07ca6cc2428616151ba7"}]},"branch":"refs/heads/master"},"411faf1d857ffbe5600c0632ac86401b62806bf5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-11 17:05:12.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/47/712447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/47/712447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/47/712447/2"}}},"commit":{"parents":[{"commit":"c93d49aa95a6973bd4fca34a8a839310a109eb4b","subject":"Add `--insecure` and `--tls-certificates` options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/c93d49aa95a6973bd4fca34a8a839310a109eb4b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-11 13:57:56.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-11 16:58:25.000000000","tz":60},"subject":"Add non-default Redfish service mount location","message":"Add non-default Redfish service mount location\n\nThe `--service-endpoint` option can now contain a path\nto Redfish API mount point. If not set, it defaults\nto sushy default which is `redfish/v1`.\n\nChange-Id: Ic7737b92c70af738c7998e412cb59d149418542e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/411faf1d857ffbe5600c0632ac86401b62806bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/411faf1d857ffbe5600c0632ac86401b62806bf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
