)]}'
{"id":"openstack%2Fsushy-cli~713418","triplet_id":"openstack%2Fsushy-cli~master~I0d427f13d509d59b742be32ef3aeee061a7de662","project":"openstack/sushy-cli","branch":"master","topic":"01-skin-the-fish","hashtags":[],"change_id":"I0d427f13d509d59b742be32ef3aeee061a7de662","subject":"Add `--show-traffic` feature to aid Redfish diagnostics","status":"MERGED","created":"2020-03-17 12:03:31.000000000","updated":"2020-04-06 15:00:52.000000000","submitted":"2020-04-06 14:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713418-1586185079623-e8a3316c","meta_rev_id":"fc3059fc7c6228b33485bae8a1ba23f50e08575f","_number":713418,"virtual_id_number":713418,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 14:57:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2020-04-06 14:45:23.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-06 14:43:32.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":1,"date":"2020-04-06 14:45:23.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 14:43:32.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-04-06 14:45:23.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-04-06 14:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b23fbee10532680b2dc57f9fcf69e9558c012bb","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-17 12:03:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"239adb72ff12b8c2517d8aa0656879c73009b7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 12:16:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/095eca2cfdc445449e15c4b9c81df617 : FAILURE in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cb447415e8a4b3d8add7ccc97e163dd : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128c7cc6b8394574aa439de1c54d454a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d45e7680234cb2b904263012ad012c : FAILURE in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5306f5e307ff41c98e71e05b24f17d64 : FAILURE in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5414925a0e44c7589d8aefa97266b12 : FAILURE in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50255dd593c4df8b69e2f0cab57186f : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8dfa85e2ea14850bf9cb8cef73c48f6 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ee75a717b0ad4206bacea408e713ca2424d8dff6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-18 16:33:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9ab7cd60968f716d10c113672caf3326b50d97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 16:44:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72474e143831421d9c72aebb1afcce3f : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bf5a6192c0b4454924ddf51a61373d4 : FAILURE in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23bb5899eb544fb94774f15e7ef86be : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5495eaa9a8ff47f19e3cc53ed05c470c : FAILURE in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb6db13ccd244e06b482ea827e3e9bc8 : FAILURE in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fd56010e55b4669acf530a9bd684b16 : FAILURE in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c47b4e278844d194ddb5aff774f7d7 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edd48319f2b14579a03818207dddd170 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"31b4052eac66417907a669779fbd3893ba477cd3","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-24 18:04:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba19c0f69fc07deb0a133aa13d5029713bcbb61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:17:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40f4470e02e641c1b7b142a945ea6a4d : FAILURE in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6dbe449328b4424977db11cff199e80 : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4edff09190df4b83a63afdcb6bca218e : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b566b5f62b364171b160167229808f62 : FAILURE in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/655eb05def7549ecbf3a111c7e6bd253 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965faedaf2a14a2896a4a2858667ef1a : FAILURE in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d9a872ba1ce4afdbecead313e44afc7 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c40ab4e53c4f1fa6876f37261b4884 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f6663290a31389b5d2cf34481792fe3782071097","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-26 15:15:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"782a2c5183f30a0949b8e0ab69cf9629408f9c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 15:28:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632f94e3deff41849bdf0dbc163790e9 : FAILURE in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/643fa3a29ce34ddd979c4d61c6c88714 : FAILURE in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf7315225d2b4ee48b1e736f1a6679ba : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f2ae7b760cf474ba8e4b446fa1fa408 : FAILURE in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f805b64a87b6435397f260a5789338e7 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/164b38eb64484d8688cd0bbbc522095d : FAILURE in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6e477a7fe0474c9825de5add2fb9a9 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31b18817de9346e0b9f92428abda1254 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3e6ecfe87685fe35187fcd732089793b8acf3adf","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-26 16:10:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"276dbc8dda687601531eaad3717f6bc5896c3f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 16:22:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d629e34894e84ab8a1c7d9dc5519fb92 : FAILURE in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2554955d10af44758a8cea334fa3af92 : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b628849a95471e830f1cd3200e09bc : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d605ad0cf634e069c251bc21aecc5a5 : FAILURE in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71247ccb12754a7481188cb5e1895a1e : FAILURE in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4745f5a46c99451989880280cf182682 : FAILURE in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f383657fa41d4695834198adb03b222f : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3ff278a07146a29cd3be9555d7f07f : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"43b7c4723deea78e6156ef1bdc25c01c59f91ac5","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-01 12:09:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"811ae4d2750e2989d30b082952fef3ffcb608531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 12:22:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3943a191328743f78752d8217c285da7 : FAILURE in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d91fad9ec0c4d7ca5540ce60ebd07fc : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1a56b8f3a74a709d7d79c3a42c2fb3 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc8c0541be2b45f4a14e650be5247b78 : FAILURE in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e794aaf5b47461895dff6c3c631ab36 : FAILURE in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f89dd8e825c54679b4fc2181f70fb960 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16164db026ab4203aa620c5f48d09b53 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece5806c579842229e83c89079ddb7c0 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ba0305332986deca436e6d2a7d44b337b3706793","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-01 12:57:34.000000000","message":"Patch Set 5:\n\nThis patch needs sushy release to pass...","accounts_in_message":[],"_revision_number":5},{"id":"5bb12f586f77863c053cd89df7998e1210d23159","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-06 10:37:22.000000000","message":"Patch Set 5:\n\nrecheck new sushy released on Apr 2","accounts_in_message":[],"_revision_number":5},{"id":"9e93ed78d7e9c25a628f696c2053b2cfc7266ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 10:50:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f4528e510de4c2980f4000a46ec9396 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2afbcb391df04a479267ccc7bce776dd : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811afa06498146f08d5a63caf05457de : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64b6f111a12043c9a3ca400f18cd667a : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8808c6daeecc4433b1e1bdfdac88a6ae : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f03879b6cd48b1b0c0d28c201a33b5 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ceafdb008148e9970131f237c76e31 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0165d54533a4eb2a97b45fccb82b495 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":5},{"id":"5fd3fa265fee7c67006302fd91a7738ae5e4e60d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-06 11:41:25.000000000","message":"Patch Set 5:\n\nI guess lower-constraints will need the new version of sushy","accounts_in_message":[],"_revision_number":5},{"id":"9f341612168f4bdeb2153b7148f4ecf12a77bb48","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-06 11:56:59.000000000","message":"Patch Set 5:\n\n\u003e I guess lower-constraints will need the new version of sushy\n\nYeah, but it needs to be released first to have a new version, right?","accounts_in_message":[],"_revision_number":5},{"id":"7f9aeb386b5a65e12bf7ebf923c30137f670e6ba","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-06 12:15:35.000000000","message":"Patch Set 5:\n\n\u003e \u003e I guess lower-constraints will need the new version of sushy\n \u003e \n \u003e Yeah, but it needs to be released first to have a new version,\n \u003e right?\n\nWasn\u0027t that made available on April 2? Version 3.2.0","accounts_in_message":[],"_revision_number":5},{"id":"6b57cb014beefb0c7ef5ef89af34ecbd11b7bfb8","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-06 13:36:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"421902d1aab6d854cb13f4e92ad5c07aa663c00e","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-06 13:37:12.000000000","message":"Patch Set 5:\n\n\u003e Wasn\u0027t that made available on April 2? Version 3.2.0\n\nRight! I missed that release! Updated.","accounts_in_message":[],"_revision_number":5},{"id":"2d8881a6e19354ad15a40659e09d5948816b7463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:51:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac814956946b4f08b7e1478e6ac08169 : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f98a997040384df28529608f3c06f837 : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a110e02451bc4de4b8be2e71946c178b : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db161bea523449d8b1662c98e3ccfcf : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07bbb39c79384977b8ac3944242d610b : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e28c53b72af3447ea9f6fd147756597e : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50c9bee8729646c7aba0ba82ea87039e : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd95a4eab0e84ffd9b4703b09d96081e : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e904b9be52a4ef9b32859c434d8a590 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2b0e910d101a1a6291eb1faa29ee6021a135e039","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-06 14:43:32.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"daed974340711d136008b983a2ad8ba8e45ec23a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-06 14:45:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b784d46e879ca4c371f0178aa4c158cd4b9e85f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:45:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1dad38e2e05f601f6bffd780780b5f0396d8e405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:57:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"45cabefbb4464fdde4b5555bb10b5a9033f89385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:57:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a77597ffa16644a58bed81c72a700c56 : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16d7d1c1ba5c4f738f6a12e2b5918858 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add254f00018446780fdbdd381da36d8 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db8ba105036d464fa4472fafc8c6e76e : SUCCESS in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12f061d2785d42829dcd47678f3287ad : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31fac5da11b47559079292cf31e0497 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6735ab2c4f74f628f79094e2c0f076b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":6},{"id":"9d48657fd2b82403ff84fbc339a0b0751080fd2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:00:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d982ae56dd48b69e63b306893bbc92 : SUCCESS in 2m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a335a283b96495fbb16981e577fd6c8 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9463bec49eb43f176e282aec1f4ede9c025cb0a3","revisions":{"e8f7ccd47cbc54d20ae9c2c7276fa3a9a2bdaf3b":{"kind":"REWORK","_number":1,"created":"2020-03-17 12:03:31.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/1"}}},"commit":{"parents":[{"commit":"1d1157dac8ff7ed510e2b973c0c2c238dbc9e80b","subject":"Add system BIOS management command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/1d1157dac8ff7ed510e2b973c0c2c238dbc9e80b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 12:03:07.000000000","tz":60},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nDepends-On: https://review.opendev.org/713315\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/e8f7ccd47cbc54d20ae9c2c7276fa3a9a2bdaf3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/e8f7ccd47cbc54d20ae9c2c7276fa3a9a2bdaf3b"}]},"branch":"refs/heads/master"},"537bfc522bba163cc505220ae36b4e07569093b1":{"kind":"REWORK","_number":2,"created":"2020-03-18 16:33:49.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/2"}}},"commit":{"parents":[{"commit":"1d1157dac8ff7ed510e2b973c0c2c238dbc9e80b","subject":"Add system BIOS management command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/1d1157dac8ff7ed510e2b973c0c2c238dbc9e80b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-18 16:11:28.000000000","tz":60},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nDepends-On: https://review.opendev.org/713315\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/537bfc522bba163cc505220ae36b4e07569093b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/537bfc522bba163cc505220ae36b4e07569093b1"}]},"branch":"refs/heads/master"},"941ee3efe42967545b316717a484c13e0a7aeda6":{"kind":"REWORK","_number":3,"created":"2020-03-26 15:15:13.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/3"}}},"commit":{"parents":[{"commit":"a08845dda48f1baf5feb453d26f60622fea6ed1d","subject":"Add system BIOS management command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/a08845dda48f1baf5feb453d26f60622fea6ed1d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-26 15:13:32.000000000","tz":60},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/941ee3efe42967545b316717a484c13e0a7aeda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/941ee3efe42967545b316717a484c13e0a7aeda6"}]},"branch":"refs/heads/master"},"c2f80eaa03997b12b6748aab5141cd47cd2c8ed2":{"kind":"REWORK","_number":4,"created":"2020-03-26 16:10:19.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/4"}}},"commit":{"parents":[{"commit":"a08845dda48f1baf5feb453d26f60622fea6ed1d","subject":"Add system BIOS management command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/a08845dda48f1baf5feb453d26f60622fea6ed1d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-26 16:09:59.000000000","tz":60},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/c2f80eaa03997b12b6748aab5141cd47cd2c8ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/c2f80eaa03997b12b6748aab5141cd47cd2c8ed2"}]},"branch":"refs/heads/master"},"b5b343b5eaa2c7b661da07711adc1789a0ed6d1d":{"kind":"REWORK","_number":5,"created":"2020-04-01 12:09:42.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/5"}}},"commit":{"parents":[{"commit":"87261f9790cf75b34f533cb8a0cf8bbb37187b47","subject":"Bump hacking to 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/87261f9790cf75b34f533cb8a0cf8bbb37187b47"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-01 12:09:20.000000000","tz":120},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/b5b343b5eaa2c7b661da07711adc1789a0ed6d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/b5b343b5eaa2c7b661da07711adc1789a0ed6d1d"}]},"branch":"refs/heads/master"},"9463bec49eb43f176e282aec1f4ede9c025cb0a3":{"kind":"REWORK","_number":6,"created":"2020-04-06 13:36:58.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/18/713418/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-cli","ref":"refs/changes/18/713418/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-cli refs/changes/18/713418/6"}}},"commit":{"parents":[{"commit":"87261f9790cf75b34f533cb8a0cf8bbb37187b47","subject":"Bump hacking to 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/87261f9790cf75b34f533cb8a0cf8bbb37187b47"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-03-17 11:32:27.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-06 13:36:36.000000000","tz":120},"subject":"Add `--show-traffic` feature to aid Redfish diagnostics","message":"Add `--show-traffic` feature to aid Redfish diagnostics\n\nAdded `--show-traffic` command-line option that makes `sushycli`\npretty-printing full HTTP exchange with the Redfish agent.\n\nThis feature can be handy for troubleshooting fishy matters.\n\nChange-Id: I0d427f13d509d59b742be32ef3aeee061a7de662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/9463bec49eb43f176e282aec1f4ede9c025cb0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-cli/commit/9463bec49eb43f176e282aec1f4ede9c025cb0a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
