)]}'
{"id":"openstack%2Fsushy-tools~640304","triplet_id":"openstack%2Fsushy-tools~master~Ia6cba82125c193f90cea9c67b6110cc4ebf264de","project":"openstack/sushy-tools","branch":"master","topic":"09-add-systems-managers-linkage","hashtags":[],"change_id":"Ia6cba82125c193f90cea9c67b6110cc4ebf264de","subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","status":"ABANDONED","created":"2019-03-01 10:16:00.000000000","updated":"2019-04-24 17:35:34.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9190ceaef727187cdf26c7189f70770b134f28d7","_number":640304,"virtual_id_number":640304,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2019-04-23 16:23:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-04 04:55:50.000000000","updated_by":{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},"reviewer":{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},"state":"REVIEWER"},{"updated":"2019-03-16 12:29:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-03-26 13:58:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-04-23 16:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d73b2113aca6e38bf41348524324bd437e610ea","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-01 10:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55f23a0cc75e68aee2be650e9bbbccb03dff7ffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 11:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/1/check/openstack-tox-pep8/15a13fd/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/1/check/openstack-tox-py27/28e1292/ : SUCCESS in 6m 30s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/1/check/openstack-tox-py35/cbe0edf/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/1/check/openstack-tox-py36/6c04d48/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/1/check/openstack-tox-lower-constraints/728825c/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/04/640304/1/check/openstack-tox-docs/65d3e05/html/ : SUCCESS in 4m 22s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/1/check/sushy-tools-tempest-ironic-partition-redfish-src/c404ac4/ : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8bc2b3ebe7d55f469217b4a6f7b16b0f0f229eb4","author":{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},"date":"2019-03-04 04:55:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30f1ac17cc429b895912fdf92e3d484af01a3921","author":{"_account_id":29209,"name":"Dhanuka Warusadura","email":"csx@tuta.io","username":"dnuka"},"date":"2019-03-04 05:17:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cffee77c0ea985011e4980929c68159290a231c4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-13 09:58:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fc1684feccd98c916939320a13cfcefb518d9cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 10:47:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/2/check/openstack-tox-pep8/386c1c0/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/2/check/openstack-tox-py27/4b11ed3/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/2/check/openstack-tox-py35/1972a3c/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/2/check/openstack-tox-py36/01bb53e/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/2/check/openstack-tox-lower-constraints/d83d515/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/04/640304/2/check/openstack-tox-docs/b22f398/html/ : SUCCESS in 5m 20s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/2/check/sushy-tools-tempest-ironic-partition-redfish-src/ed00b97/ : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"8f508603a2e64a8d0347fec7176318132570b2d5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-03-16 12:29:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90308b2456bc8c95739b4d7c6d718ef5d0febaa3","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-22 13:42:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"26eb2cb656460322e3db30923c3a82215024b505","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 14:29:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/3/check/openstack-tox-pep8/490b4c2/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/3/check/openstack-tox-py27/3810e42/ : SUCCESS in 5m 58s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/3/check/openstack-tox-py35/0fbbcb9/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/3/check/openstack-tox-py36/7e11ee1/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/3/check/openstack-tox-lower-constraints/468ac0c/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/04/640304/3/check/openstack-tox-docs/c7d0ef1/html/ : SUCCESS in 5m 04s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/3/check/sushy-tools-tempest-ironic-partition-redfish-src/c05a006/ : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":3},{"id":"8f13efee01be4160b66eb88ff7c9053bde783cba","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-22 22:03:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"bd30109e2ba4956559b22916e95bb6cf41611498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 22:53:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/4/check/openstack-tox-pep8/0b0aff4/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/4/check/openstack-tox-py27/4f803d9/ : SUCCESS in 5m 38s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/4/check/openstack-tox-py35/c4574a8/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/4/check/openstack-tox-py36/b656b7b/ : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/4/check/openstack-tox-lower-constraints/d13a0e0/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/04/640304/4/check/openstack-tox-docs/94c51a8/html/ : SUCCESS in 4m 55s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/4/check/sushy-tools-tempest-ironic-partition-redfish-src/bf69d0c/ : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":4},{"id":"14ef7de4559074b72c50d5c54e632435792f4021","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-26 06:04:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"17fa53a3088b182bda77ab214d3926f4849c1edb","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-26 07:00:27.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"92b4ec56f4f79cd00658da568f844f49dd991f5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 07:47:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/6/check/openstack-tox-pep8/93eb3de/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/6/check/openstack-tox-py27/11760c8/ : SUCCESS in 6m 53s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/6/check/openstack-tox-py35/0ac6f0c/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/6/check/openstack-tox-py36/3332fbf/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/6/check/openstack-tox-lower-constraints/c5388f0/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/04/640304/6/check/openstack-tox-docs/82646c9/html/ : SUCCESS in 6m 15s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/6/check/sushy-tools-tempest-ironic-partition-redfish-src/ff47f36/ : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f4598c4cd4a0a97b5c572820adc0ee71deb235f3","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-03-29 16:40:57.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"179f3a05eaa48097dfc56340d873a87c2a4505a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 17:27:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/7/check/openstack-tox-pep8/2e613da/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/7/check/openstack-tox-py27/f310917/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/7/check/openstack-tox-py35/15ffea1/ : SUCCESS in 6m 58s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/7/check/openstack-tox-py36/488134f/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/7/check/openstack-tox-lower-constraints/f30eac7/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/04/640304/7/check/openstack-tox-docs/0e65cc5/html/ : SUCCESS in 6m 08s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/7/check/sushy-tools-tempest-ironic-partition-redfish-src/488907f/ : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":7},{"id":"97cfeed54ef3ffa035f545fecb7d8ee2fc0a3b6f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-17 12:36:51.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"7c49f6d428b4629e800479015d6bdc1212903f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 13:39:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/8/check/openstack-tox-pep8/f828c57/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/8/check/openstack-tox-py27/0646684/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/8/check/openstack-tox-py35/107f32f/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/8/check/openstack-tox-py36/3a6c4f8/ : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/8/check/openstack-tox-lower-constraints/3be6c9a/ : SUCCESS in 9m 34s\n- openstack-tox-docs http://logs.openstack.org/04/640304/8/check/openstack-tox-docs/5e3787b/html/ : SUCCESS in 5m 31s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/8/check/sushy-tools-tempest-ironic-partition-redfish-src/3461ba9/ : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":8},{"id":"b3ca19019a9efda6e0a1aa4f56f62eb12ca0bee9","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-23 14:19:27.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"bf981fce6f3ea263e952ae233ca732459b8470d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:23:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/640304/9/check/openstack-tox-pep8/3df1a0e/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/04/640304/9/check/openstack-tox-py27/05a1aec/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/04/640304/9/check/openstack-tox-py35/d11c5f7/ : SUCCESS in 6m 38s\n- openstack-tox-py36 http://logs.openstack.org/04/640304/9/check/openstack-tox-py36/e8d3e22/ : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/640304/9/check/openstack-tox-lower-constraints/a385023/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/04/640304/9/check/openstack-tox-docs/f1d737a/html/ : SUCCESS in 4m 54s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/04/640304/9/check/sushy-tools-tempest-ironic-partition-redfish-src/10a1548/ : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":9},{"id":"9ffa14740231c8dbf37fa03e453cbf51b11fee3f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-24 17:35:34.000000000","message":"Abandoned\n\nPerhaps a better approach is to link resources up at the higher level.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"86c1579bb717ecf2f805847c31c2ee64aa63b7e4","revisions":{"ac85c9d9b13194a23317618fce7177e08aeb6d13":{"kind":"REWORK","_number":1,"created":"2019-03-01 10:16:00.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/1"}}},"commit":{"parents":[{"commit":"3fb99781dc1c3c14433e68f356c894c9f960a802","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3fb99781dc1c3c14433e68f356c894c9f960a802"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-01 10:15:22.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems drivers linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems drivers linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ac85c9d9b13194a23317618fce7177e08aeb6d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ac85c9d9b13194a23317618fce7177e08aeb6d13"}]},"branch":"refs/heads/master"},"245cb9f70476e4eed8fa737d7e489674546e4547":{"kind":"REWORK","_number":2,"created":"2019-03-13 09:58:40.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/2"}}},"commit":{"parents":[{"commit":"1a25b8960a2151a178d65aaed83965868a5c978b","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1a25b8960a2151a178d65aaed83965868a5c978b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-13 09:57:12.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/245cb9f70476e4eed8fa737d7e489674546e4547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/245cb9f70476e4eed8fa737d7e489674546e4547"}]},"branch":"refs/heads/master"},"552f43116a402ecd967040435f7cd2f0c92a74b5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-22 13:42:12.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/3"}}},"commit":{"parents":[{"commit":"14aef1643d8eb6c87cbb9f5e0a9852b6e658b88f","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/14aef1643d8eb6c87cbb9f5e0a9852b6e658b88f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-22 13:41:39.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/552f43116a402ecd967040435f7cd2f0c92a74b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/552f43116a402ecd967040435f7cd2f0c92a74b5"}]},"branch":"refs/heads/master"},"3a803a70dc8b81e531173dc7bc55b9383a989d6f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-22 22:03:36.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/4"}}},"commit":{"parents":[{"commit":"a19c36bb2c6c6796d6915f569dba44cfb82bab13","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a19c36bb2c6c6796d6915f569dba44cfb82bab13"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-22 17:46:49.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3a803a70dc8b81e531173dc7bc55b9383a989d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3a803a70dc8b81e531173dc7bc55b9383a989d6f"}]},"branch":"refs/heads/master"},"10473a474540f87ae0251eff1f7e5b0ad0a9e3eb":{"kind":"REWORK","_number":5,"created":"2019-03-26 06:04:53.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/5"}}},"commit":{"parents":[{"commit":"ead25e65173f45b104084e9dbb6ef140f7ca425d","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ead25e65173f45b104084e9dbb6ef140f7ca425d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-26 06:04:03.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/10473a474540f87ae0251eff1f7e5b0ad0a9e3eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/10473a474540f87ae0251eff1f7e5b0ad0a9e3eb"}]},"branch":"refs/heads/master"},"4a624eb73086bea6b356ba9226eb359f65e84400":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-26 07:00:27.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/6"}}},"commit":{"parents":[{"commit":"af72292e84b26688e6fa970c5a25b3a02634812d","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/af72292e84b26688e6fa970c5a25b3a02634812d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-26 06:59:37.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4a624eb73086bea6b356ba9226eb359f65e84400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4a624eb73086bea6b356ba9226eb359f65e84400"}]},"branch":"refs/heads/master"},"114b2c6800f3f3f0628ab250dc20f1aadf395e8a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-29 16:40:57.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/7"}}},"commit":{"parents":[{"commit":"80e332e7dbf43ef0cd6c707cd757e65456ea57b8","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/80e332e7dbf43ef0cd6c707cd757e65456ea57b8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-03-29 16:39:37.000000000","tz":60},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/114b2c6800f3f3f0628ab250dc20f1aadf395e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/114b2c6800f3f3f0628ab250dc20f1aadf395e8a"}]},"branch":"refs/heads/master"},"92196c12f63c3523fa4574a52933610f9ee7402b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-17 12:36:51.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/8"}}},"commit":{"parents":[{"commit":"d18eaf96582222723ea765af784c0a1dde15d3e7","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d18eaf96582222723ea765af784c0a1dde15d3e7"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-17 12:36:23.000000000","tz":120},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/92196c12f63c3523fa4574a52933610f9ee7402b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/92196c12f63c3523fa4574a52933610f9ee7402b"}]},"branch":"refs/heads/master"},"86c1579bb717ecf2f805847c31c2ee64aa63b7e4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-04-23 14:19:27.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/04/640304/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/04/640304/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/04/640304/9"}}},"commit":{"parents":[{"commit":"f0b927e623852c14bf99c4a7d20bf202cf630079","subject":"Add writable Virtual Media resource mocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f0b927e623852c14bf99c4a7d20bf202cf630079"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-28 19:54:18.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-23 14:19:17.000000000","tz":120},"subject":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator","message":"Add Managers\u003c-\u003eSystems linkage in sushy-emulator\n\nRedfish BMC is based on three key components: Systems, Managers\nand Chassis. While each component has its own properties and\nbehaviours, however in some situations these components may need\nto one another.\n\nThis change adds one-to-many style linkage between each of\nthese three objects.\n\nChange-Id: Ia6cba82125c193f90cea9c67b6110cc4ebf264de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/86c1579bb717ecf2f805847c31c2ee64aa63b7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/86c1579bb717ecf2f805847c31c2ee64aa63b7e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
