)]}'
{"id":"openstack%2Fsushy-tools~650074","triplet_id":"openstack%2Fsushy-tools~master~I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd","project":"openstack/sushy-tools","branch":"master","topic":"02-http-307-redirect","hashtags":[],"change_id":"I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd","subject":"Use HTTP code 307 for redirects","status":"MERGED","created":"2019-04-04 16:48:09.000000000","updated":"2019-05-09 09:49:54.000000000","submitted":"2019-05-09 09:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650074-1557395394889-b00d5b3e","meta_rev_id":"32e1581d636b4ddd45be7e9ca6b0d32609bf055a","_number":650074,"virtual_id_number":650074,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-09 09:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-04-29 10:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-29 10:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":1,"date":"2019-05-06 14:48:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-09 08:01:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-06 14:48:15.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-09 09:49:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72e2fd8a8a4f2546fd1e3236b05500bab51c5438","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-04 16:48:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b49f15457c742df23a24ad0d9a185f967a07749","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 17:37:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/1/check/openstack-tox-pep8/f397479/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/1/check/openstack-tox-py27/c4586c0/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/1/check/openstack-tox-py35/ff42496/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/1/check/openstack-tox-py36/b23ee3d/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/1/check/openstack-tox-lower-constraints/95354e9/ : SUCCESS in 5m 59s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/1/check/build-openstack-releasenotes/74fef5c/html/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/74/650074/1/check/openstack-tox-docs/465280e/html/ : SUCCESS in 5m 41s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/1/check/sushy-tools-tempest-ironic-partition-redfish-src/bce7cd3/ : FAILURE in 41m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4b323f2eeaa8dc63504cd81975558786572fb205","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-05 07:35:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3c7692775a90cba41b337749fc294e92eb6b7641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 08:09:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/2/check/openstack-tox-pep8/dc9d3b4/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/2/check/openstack-tox-py27/3bbb636/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/2/check/openstack-tox-py35/7ff9857/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/2/check/openstack-tox-py36/1c15ec5/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/2/check/openstack-tox-lower-constraints/9779365/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/2/check/build-openstack-releasenotes/86ea2ed/html/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/74/650074/2/check/openstack-tox-docs/6083dc2/html/ : SUCCESS in 3m 52s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/2/check/sushy-tools-tempest-ironic-partition-redfish-src/4f59d19/ : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":2},{"id":"efe0c0d3f50d30bc468aed68455185652a9c4523","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-08 09:04:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c134943a1a4f82d45f0134033503108b98024146","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-08 09:12:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee249db6051aea97a98d77ea88b586e9b2af0ee4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-08 18:15:26.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c2125b53fef9ed61fab2f862981953e487149ebd","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-08 18:31:20.000000000","message":"Patch Set 3:\n\nIt turned out that Depends-On does not readily play with infrared. Or is it my ignorance? Anyway, I went straight ahead and slid this patch under the whole cathedral of patches.","accounts_in_message":[],"_revision_number":3},{"id":"be7c954ba0bda0124d2e3498ba2383741c7a8b06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 19:13:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/3/check/openstack-tox-pep8/daec029/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/3/check/openstack-tox-py27/667d6d2/ : SUCCESS in 6m 39s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/3/check/openstack-tox-py35/afe9101/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/3/check/openstack-tox-py36/45f4410/ : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/3/check/openstack-tox-lower-constraints/84704ce/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/3/check/build-openstack-releasenotes/0a8e0df/html/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/74/650074/3/check/openstack-tox-docs/070cab4/html/ : SUCCESS in 6m 28s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/3/check/sushy-tools-tempest-ironic-partition-redfish-src/d617032/ : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":3},{"id":"847300a200052e4d49118a7521cc445050c11d64","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-08 20:40:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fb62f347839942f55f282b5f11e97af92d1c4745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 23:42:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/4/check/openstack-tox-pep8/99825fc/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/4/check/openstack-tox-py27/97beb92/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/4/check/openstack-tox-py35/df74c08/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/4/check/openstack-tox-py36/861be8d/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/4/check/openstack-tox-lower-constraints/04c2359/ : SUCCESS in 7m 53s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/4/check/build-openstack-releasenotes/f214bf8/html/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/74/650074/4/check/openstack-tox-docs/b746515/html/ : SUCCESS in 5m 30s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/4/check/sushy-tools-tempest-ironic-partition-redfish-src/ceaec16/ : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9d970e274c022e6d11b3eafbb176b29a81f1a5d1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-09 08:01:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"41a066851b6e1b90e26bbfcb53ba1f1616834836","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-09 08:33:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"51f3c289663fc90756f99b6bf1948f0c9bdb0b77","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-18 09:11:44.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"188a30be9a991f45625e40b3ca2f5a39e3968fd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 10:29:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/5/check/openstack-tox-pep8/73513e9/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/5/check/openstack-tox-py27/9597df6/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/5/check/openstack-tox-py35/153fb38/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/5/check/openstack-tox-py36/5e3064e/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/5/check/openstack-tox-lower-constraints/c7e214b/ : SUCCESS in 6m 27s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/5/check/build-openstack-releasenotes/2c19ac7/html/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/74/650074/5/check/openstack-tox-docs/7dadcde/html/ : SUCCESS in 4m 57s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/5/check/sushy-tools-tempest-ironic-partition-redfish-src/9917789/ : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":5},{"id":"53cef0311a09209b218661f8a9930315a446b62a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-18 11:04:55.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ae263add8e311badb705616c38a8b3f494293ab2","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-23 11:28:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"665449324da60af2b519d8c8a668d8483117e40b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 14:40:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/6/check/openstack-tox-pep8/00b1932/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/6/check/openstack-tox-py27/a41441f/ : SUCCESS in 6m 32s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/6/check/openstack-tox-py35/a85e628/ : SUCCESS in 6m 59s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/6/check/openstack-tox-py36/d3f15ad/ : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/6/check/openstack-tox-lower-constraints/79ab56b/ : SUCCESS in 8m 21s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/6/check/build-openstack-releasenotes/5af54ce/html/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/74/650074/6/check/openstack-tox-docs/0a00927/html/ : SUCCESS in 6m 09s\n- sushy-tools-tempest-ironic-partition-redfish-src finger://ze02.openstack.org/058bdf36a1b94451b1a153ad6195eabb : RETRY_LIMIT in 36m 36s","accounts_in_message":[],"_revision_number":6},{"id":"120189074120bddc577fdeb057278519a2f04468","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-24 10:10:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"160b3837224c4002369cd2e8e8d438378a83ffe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 10:52:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/7/check/openstack-tox-pep8/5ef6b44/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/7/check/openstack-tox-py27/968b9e7/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/7/check/openstack-tox-py35/c90dc67/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/7/check/openstack-tox-py36/a5e0a71/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/7/check/openstack-tox-lower-constraints/29972a2/ : SUCCESS in 6m 02s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/7/check/build-openstack-releasenotes/a8e1087/html/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/74/650074/7/check/openstack-tox-docs/4cf78c8/html/ : SUCCESS in 4m 38s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/7/check/sushy-tools-tempest-ironic-partition-redfish-src/dc8c22e/ : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":7},{"id":"19d87398e5181de01b38aa5aaa2dc0cd28f3619c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 12:17:17.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2531f43cd94afbd9da4690e9143036126082282a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-29 10:00:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"706547beb97070727e61f080f45f1ff3602e11ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 10:46:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/8/check/openstack-tox-pep8/ce9dab0/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/8/check/openstack-tox-py27/813e0fe/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/8/check/openstack-tox-py35/fa13e79/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/8/check/openstack-tox-py36/6e02fad/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/8/check/openstack-tox-lower-constraints/3839fb4/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/8/check/build-openstack-releasenotes/be6d7a2/html/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/74/650074/8/check/openstack-tox-docs/f767899/html/ : SUCCESS in 4m 20s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/8/check/sushy-tools-tempest-ironic-partition-redfish-src/c599c9e/ : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":8},{"id":"05bb0cf832cc37f16fa0484218a81199954efc91","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-06 14:48:15.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c2ca46811df25d704d73730c1abd817c25d2bc4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:00:00.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a55228e48cd257d6a80b8e8619338253c19f9a78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:49:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"107323b613fe3d75e3f6ee412ca4fac05b96a1af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:49:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/650074/8/gate/openstack-tox-pep8/28f3ec2/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/74/650074/8/gate/openstack-tox-py27/5a42ef1/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/74/650074/8/gate/openstack-tox-py35/ef5d93d/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/74/650074/8/gate/openstack-tox-py36/944e8c7/ : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/650074/8/gate/openstack-tox-lower-constraints/00113d2/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/74/650074/8/gate/build-openstack-releasenotes/48f15a9/html/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/74/650074/8/gate/openstack-tox-docs/944ab54/html/ : SUCCESS in 3m 53s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/74/650074/8/gate/sushy-tools-tempest-ironic-partition-redfish-src/23bce06/ : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"adc16921b4ace7ca667a14b96eab62594fb4ed88","revisions":{"5518709cbaf04b160fc48cbbc8364a1c72c80cce":{"kind":"REWORK","_number":1,"created":"2019-04-04 16:48:09.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/1"}}},"commit":{"parents":[{"commit":"ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f","subject":"Fix lower-constraint deps handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:20:05.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5518709cbaf04b160fc48cbbc8364a1c72c80cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5518709cbaf04b160fc48cbbc8364a1c72c80cce"}]},"branch":"refs/heads/master"},"62acf2bd280fd7a04f0713043b71ba732f301206":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-05 07:35:03.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/2"}}},"commit":{"parents":[{"commit":"ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f","subject":"Fix lower-constraint deps handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-05 07:34:15.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nDepends-On: https://review.openstack.org/#/c/642830/\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/62acf2bd280fd7a04f0713043b71ba732f301206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/62acf2bd280fd7a04f0713043b71ba732f301206"}]},"branch":"refs/heads/master"},"e9eef38f94b87a6d4bfe3b6dbf976e14853b1cda":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-04-08 18:15:26.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/3"}}},"commit":{"parents":[{"commit":"ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f","subject":"Fix lower-constraint deps handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ae5065cc56f0ef06b66ae780f5d6fea2f9f8305f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-08 18:13:41.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e9eef38f94b87a6d4bfe3b6dbf976e14853b1cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e9eef38f94b87a6d4bfe3b6dbf976e14853b1cda"}]},"branch":"refs/heads/master"},"1df36ffed9fb2e5344e65c8b88147e53cf76d946":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-08 20:40:40.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/4"}}},"commit":{"parents":[{"commit":"a90ec01b34d8f1431c64acccaea4779df5e3067a","subject":"Create libvirt boot loader element if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a90ec01b34d8f1431c64acccaea4779df5e3067a"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-08 20:38:35.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1df36ffed9fb2e5344e65c8b88147e53cf76d946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1df36ffed9fb2e5344e65c8b88147e53cf76d946"}]},"branch":"refs/heads/master"},"48871b0dc7c35ab6569bee5b6284461a923a6ca5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-18 09:11:44.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/5"}}},"commit":{"parents":[{"commit":"13cbfa692720ed9597401784fd317afb7e2aad84","subject":"Create libvirt boot loader element if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/13cbfa692720ed9597401784fd317afb7e2aad84"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-18 09:11:04.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/48871b0dc7c35ab6569bee5b6284461a923a6ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/48871b0dc7c35ab6569bee5b6284461a923a6ca5"}]},"branch":"refs/heads/master"},"652c41972e1fabac208ac25a6dbde788cfc41599":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-04-23 11:28:07.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/6"}}},"commit":{"parents":[{"commit":"c5b9c1bd949e36872cb1ae4f83619ae02abced5d","subject":"Create libvirt boot loader element if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c5b9c1bd949e36872cb1ae4f83619ae02abced5d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-23 11:26:48.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/652c41972e1fabac208ac25a6dbde788cfc41599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/652c41972e1fabac208ac25a6dbde788cfc41599"}]},"branch":"refs/heads/master"},"fe07ceefbbfa2f8957927cedef800471de4dc170":{"kind":"NO_CHANGE","_number":7,"created":"2019-04-24 10:10:31.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/7"}}},"commit":{"parents":[{"commit":"c5b9c1bd949e36872cb1ae4f83619ae02abced5d","subject":"Create libvirt boot loader element if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c5b9c1bd949e36872cb1ae4f83619ae02abced5d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-24 10:10:18.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fe07ceefbbfa2f8957927cedef800471de4dc170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fe07ceefbbfa2f8957927cedef800471de4dc170"}]},"branch":"refs/heads/master"},"adc16921b4ace7ca667a14b96eab62594fb4ed88":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-29 10:00:23.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/74/650074/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/650074/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/650074/8"}}},"commit":{"parents":[{"commit":"caa0593a9cca1b20de52d47a407eaa711a65e4dd","subject":"Create libvirt boot loader element if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/caa0593a9cca1b20de52d47a407eaa711a65e4dd"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-04 16:19:56.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-29 10:00:10.000000000","tz":120},"subject":"Use HTTP code 307 for redirects","message":"Use HTTP code 307 for redirects\n\nChanged HTTP redirect code from 302 to 307 to ensure the\noriginal HTTP method is left intact. Otherwise some clients\nmay change PUT/POST to GET on redirect and effectively fail\nto perform the change they intended.\n\nChange-Id: I32613e1ebb4a2eaba33c5884ddbd3ec7cf77d6fd\nStory: 2005370\nTask: 30346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/adc16921b4ace7ca667a14b96eab62594fb4ed88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/adc16921b4ace7ca667a14b96eab62594fb4ed88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
