)]}'
{"id":"openstack%2Fsushy-tools~653448","triplet_id":"openstack%2Fsushy-tools~master~If62e93e8663ae8073a957b096a9b6b3c6626e99e","project":"openstack/sushy-tools","branch":"master","topic":"15-fix-tests-layout","hashtags":[],"change_id":"If62e93e8663ae8073a957b096a9b6b3c6626e99e","subject":"Rearrange unit tests layout","status":"ABANDONED","created":"2019-04-17 14:27:18.000000000","updated":"2019-04-24 15:40:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01f7b1bdbd6667e11051f758a536bd843aa37567","_number":653448,"virtual_id_number":653448,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-23 21:47:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 21:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5af832a2a00298bb660ec6f2dca34e017ad4c31","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-17 14:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"812d86825e5f60b2bdc5ba92ac6e627323e1a097","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 14:53:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/653448/1/check/openstack-tox-pep8/49cd4e1/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/48/653448/1/check/openstack-tox-py27/43b38dc/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/48/653448/1/check/openstack-tox-py35/2e01c5b/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/48/653448/1/check/openstack-tox-py36/96ed03a/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/653448/1/check/openstack-tox-lower-constraints/1f716c0/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/48/653448/1/check/openstack-tox-docs/2fd6f3c/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"56987fd8cd2d9a5b46abd7df6cde3b43ed5b9513","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-23 14:21:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0384e978187e187e8e94d21a04f51c5c94cf09da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:47:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/653448/2/check/openstack-tox-pep8/c9b17d2/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/48/653448/2/check/openstack-tox-py27/d7a9187/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/48/653448/2/check/openstack-tox-py35/54d12e2/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/48/653448/2/check/openstack-tox-py36/9e0d7c5/ : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/653448/2/check/openstack-tox-lower-constraints/e489689/ : SUCCESS in 9m 01s\n- openstack-tox-docs http://logs.openstack.org/48/653448/2/check/openstack-tox-docs/99b30ac/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"485c281c236042d1cc5de0b830f679e91c737abb","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-04-24 15:40:25.000000000","message":"Abandoned\n\nMerged into [1].\n\n1. https://review.opendev.org/638682","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f84382617549d517911d38c1f04735dcc61dc859","revisions":{"994d73feecc9eed2a62578478cdb03b2e760c3b5":{"kind":"REWORK","_number":1,"created":"2019-04-17 14:27:18.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/48/653448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/48/653448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/1"}}},"commit":{"parents":[{"commit":"bd1365f78a3b1e2e7633192c0adbebe3fee2d8cb","subject":"Fix Flask config load under WSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/bd1365f78a3b1e2e7633192c0adbebe3fee2d8cb"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-17 14:26:07.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-17 14:26:07.000000000","tz":120},"subject":"Rearrange unit tests layout","message":"Rearrange unit tests layout\n\nReshuffles unit test files in a way to match the layout of the\nsource files being tested. This might be useful to avoid name\ncollision in the future.\n\nChange-Id: If62e93e8663ae8073a957b096a9b6b3c6626e99e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/994d73feecc9eed2a62578478cdb03b2e760c3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/994d73feecc9eed2a62578478cdb03b2e760c3b5"}]},"branch":"refs/heads/master"},"f84382617549d517911d38c1f04735dcc61dc859":{"kind":"REWORK","_number":2,"created":"2019-04-23 14:21:09.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/48/653448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/48/653448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/48/653448/2"}}},"commit":{"parents":[{"commit":"fc554f3407ffeaa47db08c8c21833862591cb909","subject":"Fix Flask config load under WSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fc554f3407ffeaa47db08c8c21833862591cb909"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-17 14:26:07.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-04-23 14:20:58.000000000","tz":120},"subject":"Rearrange unit tests layout","message":"Rearrange unit tests layout\n\nReshuffles unit test files in a way to match the layout of the\nsource files being tested. This might be useful to avoid name\ncollision in the future.\n\nChange-Id: If62e93e8663ae8073a957b096a9b6b3c6626e99e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f84382617549d517911d38c1f04735dcc61dc859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f84382617549d517911d38c1f04735dcc61dc859"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
