)]}'
{"id":"openstack%2Fsushy-tools~656231","triplet_id":"openstack%2Fsushy-tools~master~Ieaf0fa5e2db42bd742ff2117808fdec0668354eb","project":"openstack/sushy-tools","branch":"master","hashtags":[],"change_id":"Ieaf0fa5e2db42bd742ff2117808fdec0668354eb","subject":"update git.openstack.org to opendev","status":"MERGED","created":"2019-04-29 02:47:42.000000000","updated":"2019-04-29 09:32:56.000000000","submitted":"2019-04-29 09:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656231-1556530376352-0411a076","meta_rev_id":"7fbb13dcd9a5ec89c2e17c57edded0d97b184db3","_number":656231,"virtual_id_number":656231,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-04-29 09:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-04-29 08:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-04-29 08:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2019-04-29 08:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 08:52:43.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-04-29 08:57:20.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-04-29 09:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31638f84b59d1de91d7b57ae6be0e4ed8126652b","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-29 02:47:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39dcdb6052072ac2aa7f7b92f450c2a075b13604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 03:21:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/31/656231/1/check/test-release-openstack/dbc7223/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/31/656231/1/check/openstack-tox-pep8/25761a6/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/31/656231/1/check/openstack-tox-py27/6d392da/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/31/656231/1/check/openstack-tox-py35/be36efb/ : SUCCESS in 6m 58s\n- openstack-tox-py36 http://logs.openstack.org/31/656231/1/check/openstack-tox-py36/449bbaa/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/656231/1/check/openstack-tox-lower-constraints/53e6853/ : SUCCESS in 4m 53s\n- build-openstack-releasenotes http://logs.openstack.org/31/656231/1/check/build-openstack-releasenotes/bf79ae4/html/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/31/656231/1/check/openstack-tox-docs/0beaa6e/html/ : SUCCESS in 4m 05s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/31/656231/1/check/sushy-tools-tempest-ironic-partition-redfish-src/cb9e5fc/ : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3a85c32d729e57c09b066a883cbb5805785d8d01","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-29 08:52:43.000000000","message":"Patch Set 1: Code-Review+2\n\nlooks good, thanks","accounts_in_message":[],"_revision_number":1},{"id":"42b30bed1c4c4352a8479dccd62f95e48c071d82","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-04-29 08:57:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"800f61309687a06aae6ff2c3a50f32dd230f7b3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 08:57:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4037aa5423043afb35c7808ee9ac65366a93816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:32:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/31/656231/1/gate/test-release-openstack/1547e40/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/31/656231/1/gate/openstack-tox-pep8/8a10643/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/31/656231/1/gate/openstack-tox-py27/05c3e79/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/31/656231/1/gate/openstack-tox-py35/e3113fa/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/31/656231/1/gate/openstack-tox-py36/7029d25/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/656231/1/gate/openstack-tox-lower-constraints/d79dc7d/ : SUCCESS in 5m 14s\n- build-openstack-releasenotes http://logs.openstack.org/31/656231/1/gate/build-openstack-releasenotes/d784d26/html/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/31/656231/1/gate/openstack-tox-docs/02648af/html/ : SUCCESS in 4m 25s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/31/656231/1/gate/sushy-tools-tempest-ironic-partition-redfish-src/a62018a/ : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7fbb13dcd9a5ec89c2e17c57edded0d97b184db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:32:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6222c2914e42b7b576c0089c8a78d4c6cfde5c38","revisions":{"6222c2914e42b7b576c0089c8a78d4c6cfde5c38":{"kind":"REWORK","_number":1,"created":"2019-04-29 02:47:42.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/31/656231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/31/656231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/31/656231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/31/656231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/31/656231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/31/656231/1"}}},"commit":{"parents":[{"commit":"023261045fd2a6edf1219645d7e0a831cb96f3fe","subject":"Merge \"Ensure non-empty libvirt \u003cloader/\u003e tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/023261045fd2a6edf1219645d7e0a831cb96f3fe"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-29 02:47:18.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-29 02:47:38.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Ieaf0fa5e2db42bd742ff2117808fdec0668354eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6222c2914e42b7b576c0089c8a78d4c6cfde5c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6222c2914e42b7b576c0089c8a78d4c6cfde5c38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
