)]}'
{"id":"openstack%2Fsushy-tools~656792","triplet_id":"openstack%2Fsushy-tools~master~I32c7926865a4817363a5846f155b01b88ea3383c","project":"openstack/sushy-tools","branch":"master","topic":"10-add-vmedia-resource","hashtags":[],"change_id":"I32c7926865a4817363a5846f155b01b88ea3383c","subject":"Add Virtual Media support","status":"MERGED","created":"2019-05-02 15:07:34.000000000","updated":"2019-08-06 07:07:45.000000000","submitted":"2019-08-06 07:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656792-1565075265965-520616fe","meta_rev_id":"a89f2f9aa6a5619c915491d0e2c8eb10c82cbe81","_number":656792,"virtual_id_number":656792,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-08-06 07:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-07-22 18:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-07-17 09:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":1,"date":"2019-07-22 18:59:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 15:33:24.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-07-17 09:55:00.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-07-22 18:59:12.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-06 07:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb05046b6c98d4afbe8ece90f4cadd438a40ccb6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-02 15:07:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"379e0e8e20e99ead33a3c98380bb09550beaceae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:55:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/1/check/openstack-tox-pep8/1c640be/ : FAILURE in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/1/check/openstack-tox-py27/5ffed04/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/1/check/openstack-tox-py35/35561da/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/1/check/openstack-tox-py36/9243a3f/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/1/check/openstack-tox-lower-constraints/68865bd/ : SUCCESS in 5m 47s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/1/check/build-openstack-releasenotes/0856fe7/html/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/92/656792/1/check/openstack-tox-docs/7043237/html/ : SUCCESS in 4m 09s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/1/check/sushy-tools-tempest-ironic-partition-redfish-src/011c93e/ : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a45b693826c059c7e6beea8937079b962402f2bf","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-02 17:40:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a49d69ca741f36bfc33dc4be37f8038558cad23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 18:22:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/2/check/openstack-tox-pep8/2d6684b/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/2/check/openstack-tox-py27/f6c6444/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/2/check/openstack-tox-py35/857eb28/ : FAILURE in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/2/check/openstack-tox-py36/f5787a2/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/2/check/openstack-tox-lower-constraints/0abd43a/ : SUCCESS in 5m 17s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/2/check/build-openstack-releasenotes/995fa51/html/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/92/656792/2/check/openstack-tox-docs/91b3e30/html/ : SUCCESS in 8m 51s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/2/check/sushy-tools-tempest-ironic-partition-redfish-src/23f3e9a/ : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6c9987bef61b12bb58f612bbc86c5281515bd03e","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-02 18:29:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24b7a318781b1ef00775b245c44c55e88a142ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 19:24:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/3/check/openstack-tox-pep8/c8008bb/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/3/check/openstack-tox-py27/c171a85/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/3/check/openstack-tox-py35/4611523/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/3/check/openstack-tox-py36/e9bc0b0/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/3/check/openstack-tox-lower-constraints/bb72f61/ : SUCCESS in 5m 17s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/3/check/build-openstack-releasenotes/9900cc9/html/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/92/656792/3/check/openstack-tox-docs/e63134d/html/ : SUCCESS in 4m 19s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/3/check/sushy-tools-tempest-ironic-partition-redfish-src/e677ab0/ : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":3},{"id":"33e06dccadf13963dfc305056efdf7c8863b11dc","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-03 09:37:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"345fbb3094e44b763ccdfa6d0bc26ebb9b9b8eba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 10:23:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/4/check/openstack-tox-pep8/6d5c5e3/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/4/check/openstack-tox-py27/630c8d8/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/4/check/openstack-tox-py35/6385762/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/4/check/openstack-tox-py36/9d360c0/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/4/check/openstack-tox-lower-constraints/e8e4532/ : SUCCESS in 7m 49s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/4/check/build-openstack-releasenotes/e7a5a87/html/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/92/656792/4/check/openstack-tox-docs/534ae1b/html/ : SUCCESS in 3m 57s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/4/check/sushy-tools-tempest-ironic-partition-redfish-src/430dcb2/ : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":4},{"id":"8224e7e1d5a0d74ba20006888269cba8a1293bc6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-03 12:22:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cbefdc0526281f62ef3a23d0ad5b7bb551676814","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 13:06:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/5/check/openstack-tox-pep8/2f92f42/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/5/check/openstack-tox-py27/8f53c39/ : FAILURE in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/5/check/openstack-tox-py35/851e7f3/ : FAILURE in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/5/check/openstack-tox-py36/ff046c5/ : FAILURE in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/5/check/openstack-tox-lower-constraints/c073bea/ : FAILURE in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/5/check/build-openstack-releasenotes/99133f8/html/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/92/656792/5/check/openstack-tox-docs/faeddd5/html/ : SUCCESS in 6m 18s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/5/check/sushy-tools-tempest-ironic-partition-redfish-src/eda59e6/ : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":5},{"id":"cbbe6b6284ed58aca82cb671536bd83ff7b34f35","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-04 08:10:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c6501baf581406bbaef31af93dd4255e38b861c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 08:37:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/6/check/openstack-tox-pep8/d27b530/ : FAILURE in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/6/check/openstack-tox-py27/01292b2/ : FAILURE in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/6/check/openstack-tox-py35/cbbf73b/ : FAILURE in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/6/check/openstack-tox-py36/be24236/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/6/check/openstack-tox-lower-constraints/01b458d/ : FAILURE in 5m 07s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/6/check/build-openstack-releasenotes/6bde8d3/html/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/92/656792/6/check/openstack-tox-docs/e63cbe5/html/ : SUCCESS in 3m 40s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/6/check/sushy-tools-tempest-ironic-partition-redfish-src/9a14de5/ : FAILURE in 25m 54s","accounts_in_message":[],"_revision_number":6},{"id":"3d27a9a0c3655428b217b9f6a5e4faff53eb3be6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-04 15:31:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"4cf0a9354a7440eb56402ba8dfb8dafcbd98beae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:25:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/7/check/openstack-tox-pep8/2ddc205/ : FAILURE in 8m 54s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/7/check/openstack-tox-py27/7f1ae17/ : FAILURE in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/7/check/openstack-tox-py35/ced190b/ : FAILURE in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/7/check/openstack-tox-py36/79579f6/ : FAILURE in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/7/check/openstack-tox-lower-constraints/c8e0023/ : SUCCESS in 6m 15s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/7/check/build-openstack-releasenotes/fc607c0/html/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/92/656792/7/check/openstack-tox-docs/4af359d/html/ : SUCCESS in 5m 33s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/7/check/sushy-tools-tempest-ironic-partition-redfish-src/be84bd9/ : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":7},{"id":"4a9ca6b9c30be4ac267983a411c3b002334bef9c","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-04 20:25:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ba469476b1ebdbeb0fa32bc693b53d0a4fe78b27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 21:17:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/8/check/openstack-tox-pep8/3bd7635/ : FAILURE in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/8/check/openstack-tox-py27/cda05c1/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/8/check/openstack-tox-py35/4efe030/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/8/check/openstack-tox-py36/012e6ce/ : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/8/check/openstack-tox-lower-constraints/1d259df/ : SUCCESS in 5m 38s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/8/check/build-openstack-releasenotes/b1bc18f/html/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/92/656792/8/check/openstack-tox-docs/3cb17a5/html/ : SUCCESS in 6m 17s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/8/check/sushy-tools-tempest-ironic-partition-redfish-src/3436e31/ : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":8},{"id":"5c0bce4073fcd892760e5b368babcb4105f99b27","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-05 08:05:57.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"49a9ff21f95c7ae07ba3eb84b3b3b4e8757f0d6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 08:43:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/9/check/openstack-tox-pep8/45ba8e6/ : FAILURE in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/9/check/openstack-tox-py27/a12d3a0/ : SUCCESS in 7m 31s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/9/check/openstack-tox-py35/998c376/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/9/check/openstack-tox-py36/ee9c367/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/9/check/openstack-tox-lower-constraints/bd32316/ : SUCCESS in 7m 57s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/9/check/build-openstack-releasenotes/1a2b062/html/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/92/656792/9/check/openstack-tox-docs/5ad6501/html/ : SUCCESS in 6m 29s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/9/check/sushy-tools-tempest-ironic-partition-redfish-src/9351282/ : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":9},{"id":"0cc0a6d3d6416398f858331e060c0176764db9ed","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-05 18:04:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"128c948d472445687419f2f56a9ea57acf2b5c9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 18:52:48.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/10/check/openstack-tox-pep8/fc15274/ : FAILURE in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/10/check/openstack-tox-py27/b401a8f/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/10/check/openstack-tox-py35/40b086b/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/10/check/openstack-tox-py36/8c1cf44/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/10/check/openstack-tox-lower-constraints/c2039ec/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/10/check/build-openstack-releasenotes/c140db3/html/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/92/656792/10/check/openstack-tox-docs/a325a7e/html/ : SUCCESS in 4m 02s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/10/check/sushy-tools-tempest-ironic-partition-redfish-src/fb6d7e0/ : SUCCESS in 46m 12s","accounts_in_message":[],"_revision_number":10},{"id":"0b5d5d3758823e1c1b97563b73ce45966048828b","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-06 09:06:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2f096bfd8fc9eb18f61a6388c777656c85190f44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 09:58:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/11/check/openstack-tox-pep8/a0a3f35/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/11/check/openstack-tox-py27/c7e962e/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/11/check/openstack-tox-py35/7ec5f21/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/11/check/openstack-tox-py36/a0a14d8/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/11/check/openstack-tox-lower-constraints/4f19781/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/11/check/build-openstack-releasenotes/ee320be/html/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/92/656792/11/check/openstack-tox-docs/9308f63/html/ : SUCCESS in 5m 43s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/11/check/sushy-tools-tempest-ironic-partition-redfish-src/1567715/ : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":11},{"id":"ed94a28882c0b6bf36574752d2978c5650dc25d3","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-07 10:36:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"407810085a8f49f7cc84fb3e2e893808390e895d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 11:29:02.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/12/check/openstack-tox-pep8/2937ce8/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/12/check/openstack-tox-py27/4944efa/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/12/check/openstack-tox-py35/a357cc3/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/12/check/openstack-tox-py36/dca5a63/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/12/check/openstack-tox-lower-constraints/bdd4efe/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/12/check/build-openstack-releasenotes/78f705f/html/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/92/656792/12/check/openstack-tox-docs/26873b5/ : FAILURE in 5m 02s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/12/check/sushy-tools-tempest-ironic-partition-redfish-src/1f90062/ : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":12},{"id":"79e9b936a311cb848d517bb03977ca1956761a5e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-07 12:49:01.000000000","message":"Patch Set 12:\n\n(6 comments)\n\ngreat start, some comments inline","accounts_in_message":[],"_revision_number":12},{"id":"e7c00f606525f4987fc9d830e65ca7de4d860f06","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-07 14:50:20.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"de8f73e686ec193ef99447a477deeccfe7cc2c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 15:41:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/13/check/openstack-tox-pep8/5d93703/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/13/check/openstack-tox-py27/3c3a806/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/13/check/openstack-tox-py35/70dc380/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/13/check/openstack-tox-py36/5431752/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/13/check/openstack-tox-lower-constraints/8763702/ : SUCCESS in 5m 30s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/13/check/build-openstack-releasenotes/86fe772/html/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/92/656792/13/check/openstack-tox-docs/6f18121/html/ : SUCCESS in 4m 25s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/13/check/sushy-tools-tempest-ironic-partition-redfish-src/6f01b91/ : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":13},{"id":"61b82986ae707932cc389f0784d69c29ec584bbe","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-07 18:08:58.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"15d40c29e68d67d6eec4f3044e96031d55321be0","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-07 18:09:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"adbc6b2bebbb549a7f6403c58e695dad4b4d6c3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:56:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/14/check/openstack-tox-pep8/bf2c055/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/14/check/openstack-tox-py27/7f34fe9/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/14/check/openstack-tox-py35/ab893da/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/14/check/openstack-tox-py36/788a742/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/14/check/openstack-tox-lower-constraints/a97d5ce/ : SUCCESS in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/14/check/build-openstack-releasenotes/fe307ac/html/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/92/656792/14/check/openstack-tox-docs/2291287/html/ : SUCCESS in 4m 02s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/14/check/sushy-tools-tempest-ironic-partition-redfish-src/fc6393e/ : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":14},{"id":"d72ffb20f15789eea6024c05164e969f09dcf251","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-13 15:33:24.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c62bc7910dd261e9ab868bc08daa4354fe5e2b70","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-13 17:18:37.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8035788c0a272e8d792f37c91fcfc0cca67ccaa2","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-14 17:47:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"664c515c469e182732ffe68b9f689c49cb96bfcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:36:33.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/15/check/openstack-tox-pep8/6d9ff14/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/15/check/openstack-tox-py27/5fdd464/ : SUCCESS in 6m 43s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/15/check/openstack-tox-py35/9e6cc38/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/15/check/openstack-tox-py36/b8f3403/ : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/15/check/openstack-tox-lower-constraints/c727bf7/ : SUCCESS in 8m 18s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/15/check/build-openstack-releasenotes/d33cbf0/html/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/92/656792/15/check/openstack-tox-docs/48fc990/html/ : SUCCESS in 7m 14s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/15/check/sushy-tools-tempest-ironic-partition-redfish-src/4d9ba7c/ : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":15},{"id":"e62c1e0c5deefe81dd0a6dbdf8e038874482bef1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-15 07:58:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f5ba6135e65263f4f3a187e9146e5aa04ba48cde","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-20 17:35:49.000000000","message":"Uploaded patch set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"31a964e48716fae7c59a263af4a54667b4df0cc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:28:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/16/check/openstack-tox-pep8/9b690c6/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/16/check/openstack-tox-py27/98f18ba/ : FAILURE in 7m 55s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/16/check/openstack-tox-py35/906ba89/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/16/check/openstack-tox-py36/495c26c/ : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/16/check/openstack-tox-lower-constraints/35ffb16/ : SUCCESS in 7m 08s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/16/check/build-openstack-releasenotes/867b119/html/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/92/656792/16/check/openstack-tox-docs/38f961b/html/ : SUCCESS in 4m 25s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/16/check/sushy-tools-tempest-ironic-partition-redfish-src/9849059/ : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":16},{"id":"f6fe6486d6d4c83556aacdfce264c3b75eadc549","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-20 19:01:55.000000000","message":"Patch Set 16:\n\nrecheck sqlite issue","accounts_in_message":[],"_revision_number":16},{"id":"e6cbb3db1914a824b68b89e84124b65bc0bba6a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 20:09:46.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/16/check/openstack-tox-pep8/5aafe8f/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/16/check/openstack-tox-py27/76c5c36/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/16/check/openstack-tox-py35/a40e05d/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/16/check/openstack-tox-py36/3221659/ : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/16/check/openstack-tox-lower-constraints/8b45061/ : SUCCESS in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/16/check/build-openstack-releasenotes/1bcbb86/html/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/92/656792/16/check/openstack-tox-docs/a0902b1/html/ : SUCCESS in 3m 55s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/16/check/sushy-tools-tempest-ironic-partition-redfish-src/2a80344/ : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":16},{"id":"38a708f6280658373ffbc4eb9205852a32e7f91b","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-05-23 13:43:12.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a28c3f5600f4d2c8868188eea4191d204d8db831","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:37:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/17/check/openstack-tox-pep8/3b591fc/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/17/check/openstack-tox-py27/6e4c3f3/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/17/check/openstack-tox-py35/712a17e/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/17/check/openstack-tox-py36/5b88ca8/ : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/17/check/openstack-tox-lower-constraints/5be10dc/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/17/check/build-openstack-releasenotes/68f1bd1/html/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/92/656792/17/check/openstack-tox-docs/956ef71/html/ : SUCCESS in 4m 59s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/17/check/sushy-tools-tempest-ironic-partition-redfish-src/c7b1794/ : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":17},{"id":"8a97354fc8be62288fc036b7daa7232dc67e8e22","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-11 10:04:45.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"f9c65a167eef830999fd6cc6772eff7c8b2f18ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 11:23:28.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/18/check/openstack-tox-pep8/9c0c7ba/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/18/check/openstack-tox-py27/776f8d2/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/18/check/openstack-tox-py35/5531bc9/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/18/check/openstack-tox-py36/6395d2b/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/18/check/openstack-tox-lower-constraints/ea12db4/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/18/check/build-openstack-releasenotes/514a190/html/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/92/656792/18/check/openstack-tox-docs/53fc52b/html/ : SUCCESS in 5m 47s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/18/check/sushy-tools-tempest-ironic-partition-redfish-src/ee432c5/ : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":18},{"id":"1d3d248f15fb0743bb2228d36c42a06eedf6b65d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-12 09:24:47.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"8e44c6bb11a7bb749a2101c285a808048247d5be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:16:04.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/19/check/openstack-tox-pep8/9c0be55/ : FAILURE in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/19/check/openstack-tox-py27/01c07f9/ : FAILURE in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/19/check/openstack-tox-py35/89a2677/ : FAILURE in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/19/check/openstack-tox-py36/c0aca9e/ : FAILURE in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/19/check/openstack-tox-lower-constraints/7601680/ : FAILURE in 6m 01s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/19/check/build-openstack-releasenotes/7673d22/html/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/92/656792/19/check/openstack-tox-docs/79dd2de/html/ : SUCCESS in 4m 26s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/19/check/sushy-tools-tempest-ironic-partition-redfish-src/97f6a0b/ : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":19},{"id":"508c6416e29dcc1a887fbe480ffa67a1ac5ebeb4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-12 14:09:42.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"21ae597089852b5a8610629bfecf55f82b677b2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:07:44.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/20/check/openstack-tox-pep8/cec874f/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/20/check/openstack-tox-py27/3f94a03/ : FAILURE in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/92/656792/20/check/openstack-tox-py35/0f717cc/ : FAILURE in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/20/check/openstack-tox-py36/5e26afd/ : FAILURE in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/20/check/openstack-tox-lower-constraints/f33b19b/ : FAILURE in 6m 11s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/20/check/build-openstack-releasenotes/5d746f9/html/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/92/656792/20/check/openstack-tox-docs/fa1e5dc/html/ : SUCCESS in 4m 15s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/20/check/sushy-tools-tempest-ironic-partition-redfish-src/f1d2bf6/ : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":20},{"id":"4b6d24b81146e374605917b384b71a07e40ae5af","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-13 10:04:37.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"a06ec348ec00ef07eb271e791c049ba7d8ff7fe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:58:00.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/21/check/openstack-tox-pep8/e9e3ebe/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/21/check/openstack-tox-py27/2532fc1/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/21/check/openstack-tox-py36/97b1ab6/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/21/check/openstack-tox-py37/4a83921/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/21/check/openstack-tox-lower-constraints/2c730fd/ : SUCCESS in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/21/check/build-openstack-releasenotes/ca2c8de/html/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/92/656792/21/check/openstack-tox-docs/5251b6f/html/ : SUCCESS in 4m 42s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/21/check/sushy-tools-tempest-ironic-partition-redfish-src/32e0c43/ : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":21},{"id":"5a6d3e4e909836e14e2653edfcc8fadd3c9693ae","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-19 15:21:09.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"2ced8603238a22fe26934c7e2758fe9491f3b46f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:26:51.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/22/check/openstack-tox-pep8/046a11c/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/22/check/openstack-tox-py27/923a5e6/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/22/check/openstack-tox-py36/ef63c96/ : FAILURE in 7m 29s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/22/check/openstack-tox-py37/0b03ce9/ : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/22/check/openstack-tox-lower-constraints/b1b3128/ : FAILURE in 6m 23s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/22/check/build-openstack-releasenotes/c00740d/html/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/92/656792/22/check/openstack-tox-docs/b4d3d5b/html/ : SUCCESS in 5m 01s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/22/check/sushy-tools-tempest-ironic-partition-redfish-src/603d8d5/ : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":22},{"id":"584e1e345049c9f569ed3bfea96597d463fda4a1","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-20 09:28:32.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"fd5ff2fb2ee648a78569dea36e59deb34468f4be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:03:37.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/23/check/openstack-tox-pep8/8e80f74/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/23/check/openstack-tox-py27/2a28804/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/23/check/openstack-tox-py36/a2ea31b/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/23/check/openstack-tox-py37/18dad93/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/23/check/openstack-tox-lower-constraints/019536d/ : SUCCESS in 7m 02s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/23/check/build-openstack-releasenotes/56af166/html/ : SUCCESS in 10m 42s\n- openstack-tox-docs http://logs.openstack.org/92/656792/23/check/openstack-tox-docs/eccf2dd/html/ : SUCCESS in 6m 13s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/23/check/sushy-tools-tempest-ironic-partition-redfish-src/96b5423/ : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":23},{"id":"61af0c9285862ab9a2bd01be617aa03e186bf06a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-20 12:04:05.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"f9feee4e27558aec90e1b2947f0b6b22f5fb58c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:57:20.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/24/check/openstack-tox-pep8/3f000bf/ : SUCCESS in 9m 16s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/24/check/openstack-tox-py27/f89edc2/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/24/check/openstack-tox-py36/8c1812c/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/24/check/openstack-tox-py37/3432742/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/24/check/openstack-tox-lower-constraints/0fb57fd/ : SUCCESS in 6m 36s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/24/check/build-openstack-releasenotes/2faec03/html/ : SUCCESS in 10m 37s\n- openstack-tox-docs http://logs.openstack.org/92/656792/24/check/openstack-tox-docs/d800741/html/ : SUCCESS in 9m 27s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/24/check/sushy-tools-tempest-ironic-partition-redfish-src/7c718a0/ : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":24},{"id":"4203ee0199b4734791d6ee507aee2896ec38eef6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-01 10:25:44.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"51add1900a0415ac807927a0c1fccfe08fee7f7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:13:02.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/25/check/openstack-tox-pep8/26b4ee6/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/25/check/openstack-tox-py27/41c1355/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/25/check/openstack-tox-py36/b41c619/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/25/check/openstack-tox-py37/7647add/ : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/25/check/openstack-tox-lower-constraints/cc7a955/ : SUCCESS in 5m 51s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/25/check/build-openstack-releasenotes/e8c2251/html/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/92/656792/25/check/openstack-tox-docs/94d8e3d/html/ : SUCCESS in 3m 26s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/25/check/sushy-tools-tempest-ironic-partition-redfish-src/3769d25/ : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":25},{"id":"340b61fa990284b6b883dc08799bae68dd13bb39","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-02 14:13:00.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"ac93eeb5b3b94062b5df157c3361944cbd9178b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 15:06:06.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/92/656792/26/check/openstack-tox-pep8/0c0aad8/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/92/656792/26/check/openstack-tox-py27/c74f219/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/92/656792/26/check/openstack-tox-py36/992944f/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/92/656792/26/check/openstack-tox-py37/490b7df/ : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/656792/26/check/openstack-tox-lower-constraints/aaeec10/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/92/656792/26/check/build-openstack-releasenotes/4d95394/html/ : SUCCESS in 3m 00s\n- openstack-tox-docs http://logs.openstack.org/92/656792/26/check/openstack-tox-docs/e640e43/html/ : SUCCESS in 2m 58s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/656792/26/check/sushy-tools-tempest-ironic-partition-redfish-src/db3017d/ : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":26},{"id":"7c976931e6bdc5683e2760b29881b1636eab2941","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-17 09:55:00.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"910f369ad81582d05b9cc7edfb8a056c65a5b5e5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-07-22 18:59:12.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"5c1538d479aa6343c545405a52a0ad8441b1615e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 06:13:26.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"d954ededb434065d61e6af29a3c2947b97802bc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 07:07:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":26},{"id":"3d32e10391b87f7b6ceb6c79f7f2ead813d21745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 07:07:45.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/92/656792/26/gate/openstack-tox-pep8/18b0780/ : SUCCESS in 4m 01s\n- openstack-tox-py27 https://logs.opendev.org/92/656792/26/gate/openstack-tox-py27/9ef57b3/ : SUCCESS in 5m 21s\n- openstack-tox-py36 https://logs.opendev.org/92/656792/26/gate/openstack-tox-py36/dc0f820/ : SUCCESS in 5m 32s\n- openstack-tox-py37 https://logs.opendev.org/92/656792/26/gate/openstack-tox-py37/4d28f51/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://logs.opendev.org/92/656792/26/gate/openstack-tox-lower-constraints/0aef44d/ : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://logs.opendev.org/92/656792/26/gate/build-openstack-releasenotes/1dbbf1c/html/ : SUCCESS in 3m 38s\n- openstack-tox-docs https://logs.opendev.org/92/656792/26/gate/openstack-tox-docs/c74af9f/html/ : SUCCESS in 4m 28s\n- sushy-tools-tempest-ironic-partition-redfish-src https://logs.opendev.org/92/656792/26/gate/sushy-tools-tempest-ironic-partition-redfish-src/c63ea7e/ : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"e1c844dfa75b5f94b033e5d1c9b896375407f06c","revisions":{"d445451753dbd235ab39624b9d67cb18f3923fe2":{"kind":"REWORK","_number":1,"created":"2019-05-02 15:07:34.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/1"}}},"commit":{"parents":[{"commit":"90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-02 15:05:03.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just volatile mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d445451753dbd235ab39624b9d67cb18f3923fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d445451753dbd235ab39624b9d67cb18f3923fe2"}]},"branch":"refs/heads/master"},"769354b7ed5b5f895ae459fc3097caac141dc650":{"kind":"REWORK","_number":2,"created":"2019-05-02 17:40:06.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/2"}}},"commit":{"parents":[{"commit":"90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-02 17:39:35.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just volatile mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/769354b7ed5b5f895ae459fc3097caac141dc650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/769354b7ed5b5f895ae459fc3097caac141dc650"}]},"branch":"refs/heads/master"},"d9c5fbd76e4c6b0527b7e4bb0ee91298476c89d3":{"kind":"REWORK","_number":3,"created":"2019-05-02 18:29:20.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/3"}}},"commit":{"parents":[{"commit":"90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-02 18:28:48.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just volatile mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d9c5fbd76e4c6b0527b7e4bb0ee91298476c89d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d9c5fbd76e4c6b0527b7e4bb0ee91298476c89d3"}]},"branch":"refs/heads/master"},"2703654d171205927dc058ffd0e50e4300f81af4":{"kind":"REWORK","_number":4,"created":"2019-05-03 09:37:27.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/4"}}},"commit":{"parents":[{"commit":"90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/90cd6f42bf2102c7cd6d31edee63eb2fff0d8bee"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-03 09:37:00.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just volatile mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2703654d171205927dc058ffd0e50e4300f81af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2703654d171205927dc058ffd0e50e4300f81af4"}]},"branch":"refs/heads/master"},"7ca273d1e790821d3b895c53c2ae9a9c003ef0c9":{"kind":"REWORK","_number":5,"created":"2019-05-03 12:22:25.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/5"}}},"commit":{"parents":[{"commit":"1b7287afd0b3d76098a0b2fb70787b44e77a7c5c","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1b7287afd0b3d76098a0b2fb70787b44e77a7c5c"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-03 12:22:16.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just volatile mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7ca273d1e790821d3b895c53c2ae9a9c003ef0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7ca273d1e790821d3b895c53c2ae9a9c003ef0c9"}]},"branch":"refs/heads/master"},"c0e5d0b111176a3de822f27dc3d5cb2a4be7603f":{"kind":"REWORK","_number":6,"created":"2019-05-04 08:10:05.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/6"}}},"commit":{"parents":[{"commit":"6dcefdb665205db84317920d0a50e587a81f07df","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6dcefdb665205db84317920d0a50e587a81f07df"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-04 08:09:50.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c0e5d0b111176a3de822f27dc3d5cb2a4be7603f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c0e5d0b111176a3de822f27dc3d5cb2a4be7603f"}]},"branch":"refs/heads/master"},"ff211fae7e634493f40a4ef1cdcda81ebd1cbe02":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-05-04 15:31:07.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/7"}}},"commit":{"parents":[{"commit":"ca7d63a82782f4082b1938b564cd4368d33328cb","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ca7d63a82782f4082b1938b564cd4368d33328cb"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-04 15:30:54.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ff211fae7e634493f40a4ef1cdcda81ebd1cbe02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ff211fae7e634493f40a4ef1cdcda81ebd1cbe02"}]},"branch":"refs/heads/master"},"71928c3faa821ad11ce33644d0472743a9b5aafc":{"kind":"REWORK","_number":8,"created":"2019-05-04 20:25:23.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/8"}}},"commit":{"parents":[{"commit":"313d08ce85548a8382056afcf374729e11a1bea8","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/313d08ce85548a8382056afcf374729e11a1bea8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-04 20:25:09.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/71928c3faa821ad11ce33644d0472743a9b5aafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/71928c3faa821ad11ce33644d0472743a9b5aafc"}]},"branch":"refs/heads/master"},"0cab8e19f043e21bc832cdf891f431accf863751":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-05 08:05:57.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/9"}}},"commit":{"parents":[{"commit":"df5fadde49d1ca85d492325ce677ea91b64dc474","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/df5fadde49d1ca85d492325ce677ea91b64dc474"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-05 08:05:43.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0cab8e19f043e21bc832cdf891f431accf863751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0cab8e19f043e21bc832cdf891f431accf863751"}]},"branch":"refs/heads/master"},"d88918138bd23a219e2a100b096faa2301b920c3":{"kind":"REWORK","_number":10,"created":"2019-05-05 18:04:31.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/10"}}},"commit":{"parents":[{"commit":"df5fadde49d1ca85d492325ce677ea91b64dc474","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/df5fadde49d1ca85d492325ce677ea91b64dc474"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-05 18:04:03.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d88918138bd23a219e2a100b096faa2301b920c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d88918138bd23a219e2a100b096faa2301b920c3"}]},"branch":"refs/heads/master"},"1072b68b10988880da9051eab9e3abfcdec6bb63":{"kind":"REWORK","_number":11,"created":"2019-05-06 09:06:19.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/11"}}},"commit":{"parents":[{"commit":"df5fadde49d1ca85d492325ce677ea91b64dc474","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/df5fadde49d1ca85d492325ce677ea91b64dc474"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-06 09:05:12.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1072b68b10988880da9051eab9e3abfcdec6bb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1072b68b10988880da9051eab9e3abfcdec6bb63"}]},"branch":"refs/heads/master"},"5c8602e61bdd412a32c15788c34ee7871d925074":{"kind":"REWORK","_number":12,"created":"2019-05-07 10:36:47.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/12"}}},"commit":{"parents":[{"commit":"b598f43a941031f56ea384a921add7222d898e0b","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b598f43a941031f56ea384a921add7222d898e0b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-07 10:36:13.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5c8602e61bdd412a32c15788c34ee7871d925074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5c8602e61bdd412a32c15788c34ee7871d925074"}]},"branch":"refs/heads/master"},"1d8bfce594636455febf0b2d6e9b8709d08a927f":{"kind":"REWORK","_number":13,"created":"2019-05-07 14:50:20.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/13"}}},"commit":{"parents":[{"commit":"b598f43a941031f56ea384a921add7222d898e0b","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b598f43a941031f56ea384a921add7222d898e0b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-07 14:49:47.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1d8bfce594636455febf0b2d6e9b8709d08a927f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1d8bfce594636455febf0b2d6e9b8709d08a927f"}]},"branch":"refs/heads/master"},"35028de8f3efd25ff3d54f20bdc2c6378afb1508":{"kind":"REWORK","_number":14,"created":"2019-05-07 18:09:34.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/14"}}},"commit":{"parents":[{"commit":"b598f43a941031f56ea384a921add7222d898e0b","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b598f43a941031f56ea384a921add7222d898e0b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-07 18:09:11.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/35028de8f3efd25ff3d54f20bdc2c6378afb1508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/35028de8f3efd25ff3d54f20bdc2c6378afb1508"}]},"branch":"refs/heads/master"},"d67838bcd2864ad3fea175a0c72498448b6ee1bb":{"kind":"REWORK","_number":15,"created":"2019-05-14 17:47:56.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/15"}}},"commit":{"parents":[{"commit":"b598f43a941031f56ea384a921add7222d898e0b","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b598f43a941031f56ea384a921add7222d898e0b"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-14 17:47:16.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d67838bcd2864ad3fea175a0c72498448b6ee1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d67838bcd2864ad3fea175a0c72498448b6ee1bb"}]},"branch":"refs/heads/master"},"3c3f7e567e2ee33703d084110b45db09a64999ed":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2019-05-20 17:35:49.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/16"}}},"commit":{"parents":[{"commit":"109a10a013fd217b3ae3a0696d04312331a5d620","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/109a10a013fd217b3ae3a0696d04312331a5d620"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-20 17:35:39.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3c3f7e567e2ee33703d084110b45db09a64999ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3c3f7e567e2ee33703d084110b45db09a64999ed"}]},"branch":"refs/heads/master"},"481509696643d4594c4a85ef0f2433cbfe9ce8c0":{"kind":"REWORK","_number":17,"created":"2019-05-23 13:43:12.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/17"}}},"commit":{"parents":[{"commit":"109a10a013fd217b3ae3a0696d04312331a5d620","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/109a10a013fd217b3ae3a0696d04312331a5d620"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-05-23 13:42:32.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/481509696643d4594c4a85ef0f2433cbfe9ce8c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/481509696643d4594c4a85ef0f2433cbfe9ce8c0"}]},"branch":"refs/heads/master"},"5701729fbea3a6866266f67bc7cd2ae2400d2959":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-06-11 10:04:45.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/18"}}},"commit":{"parents":[{"commit":"fc2a84797d3a21494db187f31927d32f1e81d1c7","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fc2a84797d3a21494db187f31927d32f1e81d1c7"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-11 10:04:34.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5701729fbea3a6866266f67bc7cd2ae2400d2959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5701729fbea3a6866266f67bc7cd2ae2400d2959"}]},"branch":"refs/heads/master"},"5a59b2b2e3b0d054570db7c70a13448f4be83931":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-06-12 09:24:47.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/19"}}},"commit":{"parents":[{"commit":"bc85fdc1d4769ec5baba327c92b0661d1b54be7c","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/bc85fdc1d4769ec5baba327c92b0661d1b54be7c"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-12 09:24:37.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5a59b2b2e3b0d054570db7c70a13448f4be83931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5a59b2b2e3b0d054570db7c70a13448f4be83931"}]},"branch":"refs/heads/master"},"f5e9ede7dea2617e1444e6618e6757c0a04ccc7e":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-06-12 14:09:42.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/20"}}},"commit":{"parents":[{"commit":"4610922ba9885a0f834e0913783eaa601086980d","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4610922ba9885a0f834e0913783eaa601086980d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-12 14:09:31.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f5e9ede7dea2617e1444e6618e6757c0a04ccc7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f5e9ede7dea2617e1444e6618e6757c0a04ccc7e"}]},"branch":"refs/heads/master"},"6f6384cf6dfd9bb69110a9380f12a54dbe8fc454":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-06-13 10:04:37.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/21"}}},"commit":{"parents":[{"commit":"6b67fc67cbc39a0db41f5b3bbd60aa1b78de5276","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6b67fc67cbc39a0db41f5b3bbd60aa1b78de5276"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-13 10:04:26.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6f6384cf6dfd9bb69110a9380f12a54dbe8fc454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6f6384cf6dfd9bb69110a9380f12a54dbe8fc454"}]},"branch":"refs/heads/master"},"62c2a5bdd1814e44835398acd083d0ffb8630fa2":{"kind":"REWORK","_number":22,"created":"2019-06-19 15:21:09.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/22"}}},"commit":{"parents":[{"commit":"92c86531d7fcdea3d6ba1ccf390d18b5ba8d6f12","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/92c86531d7fcdea3d6ba1ccf390d18b5ba8d6f12"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-19 15:20:21.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/62c2a5bdd1814e44835398acd083d0ffb8630fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/62c2a5bdd1814e44835398acd083d0ffb8630fa2"}]},"branch":"refs/heads/master"},"8f61537ce7da0e483d9721dd4a4fc5d404ebda21":{"kind":"REWORK","_number":23,"created":"2019-06-20 09:28:32.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/23"}}},"commit":{"parents":[{"commit":"c0c5dac83ef788f3995ea70174ac9852a05f5ddd","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c0c5dac83ef788f3995ea70174ac9852a05f5ddd"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-20 09:28:19.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8f61537ce7da0e483d9721dd4a4fc5d404ebda21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8f61537ce7da0e483d9721dd4a4fc5d404ebda21"}]},"branch":"refs/heads/master"},"b601fbd220ef9c862a472c9bc4a8c2ace48f5449":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-06-20 12:04:05.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/24"}}},"commit":{"parents":[{"commit":"e862a2a834e3f4d347625494683f57029b8ffa3f","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e862a2a834e3f4d347625494683f57029b8ffa3f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-20 12:03:53.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b601fbd220ef9c862a472c9bc4a8c2ace48f5449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b601fbd220ef9c862a472c9bc4a8c2ace48f5449"}]},"branch":"refs/heads/master"},"4da070b3be3acb587d0d17688e1c55cdb9affd05":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2019-07-01 10:25:44.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/25"}}},"commit":{"parents":[{"commit":"13f3bccf37d7257820f6748863480c1c21edb133","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/13f3bccf37d7257820f6748863480c1c21edb133"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-01 10:25:37.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4da070b3be3acb587d0d17688e1c55cdb9affd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4da070b3be3acb587d0d17688e1c55cdb9affd05"}]},"branch":"refs/heads/master"},"e1c844dfa75b5f94b033e5d1c9b896375407f06c":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-07-02 14:13:00.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/92/656792/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/92/656792/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/92/656792/26"}}},"commit":{"parents":[{"commit":"155fed1aeb4d1b50dc95a46f1cb771e2b74b8394","subject":"Add Indicator LED resource emulation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/155fed1aeb4d1b50dc95a46f1cb771e2b74b8394"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-02-22 23:15:31.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-02 14:12:51.000000000","tz":120},"subject":"Add Virtual Media support","message":"Add Virtual Media support\n\nThis patch introduces VirtualMedia resource.\n\nThis change implements Insert/Eject actions against the VirtualMedia\nresource. Although all changes are just persistent mocks - virtual media\nemulation is not yet backed by virtualization drivers yet.\n\nThe upcoming patches will convey virtual media calls to systems\ndrivers (e.g. libvirt and/or nova) to make VMs bootable from the\ninserted images.\n\nStory: 2005149\nTask: 29857\nChange-Id: I32c7926865a4817363a5846f155b01b88ea3383c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e1c844dfa75b5f94b033e5d1c9b896375407f06c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e1c844dfa75b5f94b033e5d1c9b896375407f06c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
