)]}'
{"id":"openstack%2Fsushy-tools~666296","triplet_id":"openstack%2Fsushy-tools~master~Ib82485c0cb65a122593308a43eedd257e1e61eab","project":"openstack/sushy-tools","branch":"master","topic":"08-add-persistance-mechanism","hashtags":[],"change_id":"Ib82485c0cb65a122593308a43eedd257e1e61eab","subject":"Add state sharing and persistence mechanism","status":"MERGED","created":"2019-06-19 13:58:21.000000000","updated":"2019-08-22 21:39:00.000000000","submitted":"2019-08-06 07:07:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666296-1565075260544-30793c9e","meta_rev_id":"671541a0b8eced21c130a30f36f09b547363f073","_number":666296,"virtual_id_number":666296,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-05 15:02:10.000000000","_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"date":"2019-07-05 09:16:38.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-08-06 07:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-07-22 14:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-08-06 06:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 21:42:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2019-08-06 06:13:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 09:16:38.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-07-22 14:04:16.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-05 21:42:44.000000000","updated_by":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"reviewer":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"state":"REVIEWER"},{"updated":"2019-08-06 06:13:13.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-08-06 07:07:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"767b478a2564aea7e4479a701e0a4497e908b7fe","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-19 13:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d8b02e0885b724e8c66c1651bd5bf0e85713f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 14:48:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/96/666296/1/check/openstack-tox-pep8/f1edcb5/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/96/666296/1/check/openstack-tox-py27/282ab7c/ : FAILURE in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/96/666296/1/check/openstack-tox-py36/c038e69/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/96/666296/1/check/openstack-tox-py37/12d2d3c/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/666296/1/check/openstack-tox-lower-constraints/73bfeab/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/96/666296/1/check/openstack-tox-docs/d0d2e0c/html/ : SUCCESS in 4m 28s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/96/666296/1/check/sushy-tools-tempest-ironic-partition-redfish-src/9d5d096/ : SUCCESS in 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9b8d873db166af2a4b993ac9419547c639d05aeb","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-19 15:18:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0243cb7a5fd915beabac97af1f393b8fab598046","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:07:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/666296/2/check/openstack-tox-pep8/4d15d4c/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/96/666296/2/check/openstack-tox-py27/b1d42b3/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/96/666296/2/check/openstack-tox-py36/7ef8876/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/96/666296/2/check/openstack-tox-py37/b958ee5/ : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/666296/2/check/openstack-tox-lower-constraints/4f89b3c/ : SUCCESS in 7m 39s\n- openstack-tox-docs http://logs.openstack.org/96/666296/2/check/openstack-tox-docs/1e37597/html/ : SUCCESS in 4m 21s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/96/666296/2/check/sushy-tools-tempest-ironic-partition-redfish-src/a3c1034/ : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":2},{"id":"60f63f8a2070f526c113c8c57fb4cefc269af69f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-20 09:28:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"080a3ed131f35458a804e563d9013f22b787d26c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/666296/3/check/openstack-tox-pep8/8527a9f/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/96/666296/3/check/openstack-tox-py27/294186e/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/96/666296/3/check/openstack-tox-py36/028a49b/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/96/666296/3/check/openstack-tox-py37/c13719e/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/666296/3/check/openstack-tox-lower-constraints/57bdcee/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/96/666296/3/check/openstack-tox-docs/e2ced59/html/ : SUCCESS in 4m 46s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/96/666296/3/check/sushy-tools-tempest-ironic-partition-redfish-src/fecda82/ : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":3},{"id":"cb3496290ddbe7101ecf4ece02f1d3ce8a9341be","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-01 10:25:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"42736a3d2012da59ed956c68b1dedb74b2589d01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:10:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/666296/4/check/openstack-tox-pep8/d300cc5/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/96/666296/4/check/openstack-tox-py27/e79c18b/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/96/666296/4/check/openstack-tox-py36/1ff64d7/ : SUCCESS in 3m 00s\n- openstack-tox-py37 http://logs.openstack.org/96/666296/4/check/openstack-tox-py37/0759815/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/666296/4/check/openstack-tox-lower-constraints/56ef3a0/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/96/666296/4/check/openstack-tox-docs/049d927/html/ : SUCCESS in 3m 03s\n- sushy-tools-tempest-ironic-partition-redfish-src http://logs.openstack.org/96/666296/4/check/sushy-tools-tempest-ironic-partition-redfish-src/ad0a39a/ : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":4},{"id":"bd96cb45b7d75bc6f1f7e790812e23cc3642559a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-05 09:16:38.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e775e0bad702d956219344bbbf579f26fe5ab329","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-07-22 14:04:16.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0c2b3fa020736775a59cb8016b2c15535da6c4a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-05 15:02:10.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2e427c26d1c83722bc609811e92bdc4eb81584e3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-05 18:51:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"58c6a37909dc7d4d59ef2d29afd84e519ac0241a","author":{"_account_id":8125,"name":"Michael Davies","email":"michael@the-davies.net","username":"mrda"},"date":"2019-08-05 21:42:44.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":4},{"id":"c2fd0c2601c5e91f3270769c82377136f1c755d0","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-08-06 06:13:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)\n\nThanks Ilya, the patch looks good to me (without looking patches in the chain). Issues below could be done in the follow up if necessary.","accounts_in_message":[],"_revision_number":4},{"id":"a6d99f3359a3dd6cf7e2bfd3863c11d2931945eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 06:13:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c6b5d2481614d9c4b122ccca1ec6d27b286ad1fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 07:07:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/96/666296/4/gate/openstack-tox-pep8/d7a48a6/ : SUCCESS in 5m 22s\n- openstack-tox-py27 https://logs.opendev.org/96/666296/4/gate/openstack-tox-py27/88b84b1/ : SUCCESS in 4m 22s\n- openstack-tox-py36 https://logs.opendev.org/96/666296/4/gate/openstack-tox-py36/8e2833f/ : SUCCESS in 5m 05s\n- openstack-tox-py37 https://logs.opendev.org/96/666296/4/gate/openstack-tox-py37/1d0ab50/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://logs.opendev.org/96/666296/4/gate/openstack-tox-lower-constraints/2b86db7/ : SUCCESS in 5m 55s\n- openstack-tox-docs https://logs.opendev.org/96/666296/4/gate/openstack-tox-docs/570be2a/html/ : SUCCESS in 4m 34s\n- sushy-tools-tempest-ironic-partition-redfish-src https://logs.opendev.org/96/666296/4/gate/sushy-tools-tempest-ironic-partition-redfish-src/32cebee/ : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":4},{"id":"dc049940ae12b467b84bf2243838674eb2eaa42a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 07:07:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"bc62237c14935a099e8866e83c7c3e95b0407e0f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-13 08:56:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ae8b05a0c4925d1b216d226d84efd34cff4c8f32","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-21 08:44:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c52556ee16823e1552de0cf5594b94d80f0b93ac","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-22 21:39:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b4cba952abf933fee9ecbb7b8782c42bf349d2e4","revisions":{"99d33e68978d4019d9cab8429930cfd1111d80b7":{"kind":"REWORK","_number":1,"created":"2019-06-19 13:58:21.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/96/666296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/96/666296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/1"}}},"commit":{"parents":[{"commit":"8bd1c355341d091d399b1adb7915d0e5b320db98","subject":"Add Chassis resource support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8bd1c355341d091d399b1adb7915d0e5b320db98"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-18 14:10:31.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-19 13:57:51.000000000","tz":120},"subject":"Add state sharing and persistence mechanism","message":"Add state sharing and persistence mechanism\n\nFor some Redfish resources, emulator needs to keep a state\ninternally. Given that dynamic emulator is a potentially multi-process\napp, states need to be shared across running emulator instances.\n\nThis change introduces persistent Python dict backed by sqlite DB.\n\nChange-Id: Ib82485c0cb65a122593308a43eedd257e1e61eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/99d33e68978d4019d9cab8429930cfd1111d80b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/99d33e68978d4019d9cab8429930cfd1111d80b7"}]},"branch":"refs/heads/master"},"35ff3a021db56d45183b4ff538abaf5ad1bc2b99":{"kind":"REWORK","_number":2,"created":"2019-06-19 15:18:28.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/96/666296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/96/666296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/2"}}},"commit":{"parents":[{"commit":"927b8f7b5a38d5d73c44a52ce91d356bdd0c82d5","subject":"Add Chassis resource support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/927b8f7b5a38d5d73c44a52ce91d356bdd0c82d5"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-18 14:10:31.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-19 15:18:18.000000000","tz":120},"subject":"Add state sharing and persistence mechanism","message":"Add state sharing and persistence mechanism\n\nFor some Redfish resources, emulator needs to keep a state\ninternally. Given that dynamic emulator is a potentially multi-process\napp, states need to be shared across running emulator instances.\n\nThis change introduces persistent Python dict backed by sqlite DB.\n\nChange-Id: Ib82485c0cb65a122593308a43eedd257e1e61eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/35ff3a021db56d45183b4ff538abaf5ad1bc2b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/35ff3a021db56d45183b4ff538abaf5ad1bc2b99"}]},"branch":"refs/heads/master"},"7c1c4888792171e3337a31c2ba6f4650d24f53d3":{"kind":"REWORK","_number":3,"created":"2019-06-20 09:28:01.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/96/666296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/96/666296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/3"}}},"commit":{"parents":[{"commit":"927b8f7b5a38d5d73c44a52ce91d356bdd0c82d5","subject":"Add Chassis resource support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/927b8f7b5a38d5d73c44a52ce91d356bdd0c82d5"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-18 14:10:31.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-20 09:27:50.000000000","tz":120},"subject":"Add state sharing and persistence mechanism","message":"Add state sharing and persistence mechanism\n\nFor some Redfish resources, emulator needs to keep a state\ninternally. Given that dynamic emulator is a potentially multi-process\napp, states need to be shared across running emulator instances.\n\nThis change introduces persistent Python dict backed by sqlite DB.\n\nChange-Id: Ib82485c0cb65a122593308a43eedd257e1e61eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7c1c4888792171e3337a31c2ba6f4650d24f53d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7c1c4888792171e3337a31c2ba6f4650d24f53d3"}]},"branch":"refs/heads/master"},"b4cba952abf933fee9ecbb7b8782c42bf349d2e4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-01 10:25:24.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/96/666296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/96/666296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/96/666296/4"}}},"commit":{"parents":[{"commit":"248e3d1353d76e3a2f3f201f8d937c5f518f0036","subject":"Add Chassis resource support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/248e3d1353d76e3a2f3f201f8d937c5f518f0036"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-06-18 14:10:31.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-01 10:25:16.000000000","tz":120},"subject":"Add state sharing and persistence mechanism","message":"Add state sharing and persistence mechanism\n\nFor some Redfish resources, emulator needs to keep a state\ninternally. Given that dynamic emulator is a potentially multi-process\napp, states need to be shared across running emulator instances.\n\nThis change introduces persistent Python dict backed by sqlite DB.\n\nChange-Id: Ib82485c0cb65a122593308a43eedd257e1e61eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b4cba952abf933fee9ecbb7b8782c42bf349d2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b4cba952abf933fee9ecbb7b8782c42bf349d2e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
