)]}'
{"id":"openstack%2Fsushy-tools~684685","triplet_id":"openstack%2Fsushy-tools~master~Id04caa4f1b0701af219cf85d6e30405ead097ec3","project":"openstack/sushy-tools","branch":"master","topic":"04-fix-listen-ip-config","hashtags":[],"change_id":"Id04caa4f1b0701af219cf85d6e30405ead097ec3","subject":"Fix emulator socket binding options","status":"MERGED","created":"2019-09-25 10:13:22.000000000","updated":"2019-11-27 12:14:00.000000000","submitted":"2019-11-27 12:11:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684685-1574856689979-ca8dfbed","meta_rev_id":"a647a3a5f3526424fe9007eb9786f992e4addb95","_number":684685,"virtual_id_number":684685,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 12:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-11-25 17:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-26 14:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 08:06:10.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-11-26 14:00:59.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-11-27 12:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bac25c0bc969fe9d316f5a1ce7a8ec02c6ba5651","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-09-25 10:13:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d373f1a340a6ae05397c3b23e40d0cfe656cb1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 11:17:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a94b16ef804c25858127e30018beab : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8a9ddc13a0d4277a637d70986b06ca3 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617fa3ef07524680b9c2c7b9f7bbb744 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43edaca85a2d4362ba7aa1f664851d38 : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2346d2f0a8de40b2ab92069d7e157d33 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b52509ce72f4880b3a299304e967d48 : SUCCESS in 7m 47s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/bf9ada7b30634777b5ee01c79aaa7fa2 : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c4f5ba79bf17fd414e9f793821a5c966778f6b65","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-09-26 08:46:38.000000000","message":"Patch Set 1:\n\nrecheck\n\nSomehow I do not see logs from the failed job at all.","accounts_in_message":[],"_revision_number":1},{"id":"8ac27129fd1d408df875c7b766bc7f2de44dcd50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:57:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe87f8597f140feb281feff3e206224 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d74ce30888a4f158fa60192fed98857 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ff02c8acd74dad9dda5fe38eb2fe0c : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a717e8e49d50465995751ceef1aea5a5 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aec6f2bfc3e47b689a1e75621dce66f : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff27000a1164d0e93779d73a5fcd9a0 : SUCCESS in 5m 44s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/169d045c937e4781b2c7873bbedbfcd4 : FAILURE in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8e935a569b71543be2e0363e967bd7a69e2dac33","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-09-30 11:11:40.000000000","message":"Patch Set 1:\n\nrecheck\nTempest failure seems unrelated.","accounts_in_message":[],"_revision_number":1},{"id":"df2d74236332d97fad318b0fd2792f8c4a950135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 12:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fc49bb615d44f4b5d66e780fe708ad : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aeae0a74e4954d5491c47846fae55341 : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebbba85163e44600833a97e73be6ebd4 : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecf4cd65fd8e4d4c90c714883ffdf47c : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ad102ffec0e4060a0389d12957e6b8a : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f67cb009f8493cad6b9f02a8bcf5ca : SUCCESS in 4m 56s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/48ccc8e9b944474c8c0699c374bcd69b : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":1},{"id":"72840e2b5067430a3b729415c0978aa300f53704","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 16:10:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e93c149403ab53db613ffecc6200d44d1b5c898e","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 16:18:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2660dc8ec2308a5100dfd59daf07728c5c28267e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:11:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56fe9c7cd1bd4802a6647184e5021a44 : SUCCESS in 2m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9ebf28e4ea04f39b68c785d252edd24 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5afb1ae987c84c7a84c490ff21ba2604 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f55833db73874b61ad437202430cede6 : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f5b0c4b2f2f4e57b2aaf3757725a7bc : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd89a89d419f461b9209ed6c46be2254 : SUCCESS in 5m 38s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/774d2d27c9514c5e84376dddb132aa92 : FAILURE in 49m 01s","accounts_in_message":[],"_revision_number":3},{"id":"6f12775ebf93a165ea34a5d808d7ed6c55baaf1f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 20:51:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4a7906c9452ffa291bcb00b235415974ab24a71a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:49:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116232c2ee6643bdbbf736b7b604c5e8 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1838912cd9254bb68d5f4133b4b4f0c9 : FAILURE in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e716b649be0548f6b9b1e58a9d0a8e06 : FAILURE in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30dd25debb0a457bbf4fd4870b970b64 : FAILURE in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88a18ba73a454229a6edc64d33c23681 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d151f000ae6418285de7ee09cb43efb : SUCCESS in 5m 46s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/192edbdf2ff248e0a22d06824a4e991f : FAILURE in 48m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1d270b8ce8e159460e0b0f204fb99cae21cec2ad","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-19 21:25:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"5176f93d7f5047ae03a7727d677b393dc04cc891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 22:16:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782b2d53ef6a4d7cbc5c8b256a4bb272 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4af467e3bfbb4d6490e65ac7fd48e736 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e5c393c445a49deaf9e5303aaeb1626 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f11ac92df6db463490bd630a6b4e3fd6 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c780917ee2e4683b6e96ab13ce3f354 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6311b1b139fd4ff3aadd258a2776797f : SUCCESS in 4m 25s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/92651da5217a48bba929fd7469d1fca8 : FAILURE in 46m 27s","accounts_in_message":[],"_revision_number":5},{"id":"a6b89821b8543e78ed3957f1ae85145d1e6f6600","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-24 11:29:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"be2b9a2fad061d403346409b4877439893562e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:22:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2995d0788f484761b1a5f34223422f0e : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1772b510791e40da8072ea9af146dc7a : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828abfdbb1ea48f3b04200cf29f239b4 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e1d146966ea41f4b1f6b71f6d9ae584 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77878998663546ffbe04dce9b1f56d6b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84d169c40b64545bf6783b7e5708a17 : SUCCESS in 4m 51s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/84363dd60d1a4f52a72ac1c72de6b60f : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":6},{"id":"7edcfe82a000928659f0622f15e9b48c35a0a065","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-19 08:06:10.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"50ba21d7ca894cdedb853b82c6ba36d6db768033","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-25 17:43:56.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"92d22da25d2dcbb1ef87b55d4a57fd574202681a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:53:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1e3b06d41a428688301ce188c051e5 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faea7898209e4ee38e998e3b2f1f7498 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5e6431d10194f06913a1ba0ec493f14 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bda8dcaabb5947158c9922924bfd8c3f : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbb89fe8e2f4547921fe85361948d69 : SUCCESS in 4m 48s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/e6f995f1ea0340d29903f5d6042fc4ad : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":7},{"id":"3b5f33c902305479194cf45068e0f9112e08845f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-26 14:00:59.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1e794cf3056935130ffc8b2438e925f7e9c03256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 14:01:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bba75e45fdd23384d0939bd11451441332084b67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 14:53:23.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"63b9f6527c4800830e9a1ef2985a01cd91e5e1c5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-27 10:32:31.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8c43301c89500dbe4fa1aa94ad773a00621848a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 11:16:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47aceaff303a45aaa8e7c96cad0b0fce : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5fcce90ff7749f6b54c46f020e3ea1d : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e168ed74bc314d24be9c96c55c92a6c9 : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c1bb015c1d94b32bfefaf880adf6c51 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeed16ce94254970983f659a689df4e0 : SUCCESS in 4m 40s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/bc6c26fc30704774b5b4a32f1623e705 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":7},{"id":"bd9138772a0be6a37bff4aa1b302468dd5d8879d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 11:22:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7a18cba2d7435f089aa75b7326b17769f1bd043f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:11:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"225184a56bdb364cbfeb7e1b2a5c89f7051aed51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:11:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e98e51b0a4f43e8808434e31a028934 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d55e6517804e4b078b695bb21d3cbb30 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8c17a57c3f042629fc860a345601914 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/926fdc72df234ea79b4c86ca2ea1df47 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d466bd994ef4dc0b7dc745f91adfbb2 : SUCCESS in 4m 55s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/584e08e200af4e9ab32820a3a0f3b0ac : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":7},{"id":"e4d8eac21074490a422dd7bfc04339963450c44e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:14:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9ccb46f30c4aadb2a23be6c2ef370c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"eb2943dfab00cb9e58eeddd97bc0d255f320120c","revisions":{"e16d9221bd4107f3ab65ca29dfe17717e6ef853d":{"kind":"REWORK","_number":1,"created":"2019-09-25 10:13:22.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/1"}}},"commit":{"parents":[{"commit":"aed2f999c70ed28b4c755867e68bddd62bb73495","subject":"Merge \"Store virtual media device changes persistently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/aed2f999c70ed28b4c755867e68bddd62bb73495"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e16d9221bd4107f3ab65ca29dfe17717e6ef853d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e16d9221bd4107f3ab65ca29dfe17717e6ef853d"}]},"branch":"refs/heads/master"},"36d4989b6589954297f42694123a2cd6c0d275ca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-14 16:10:53.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/2"}}},"commit":{"parents":[{"commit":"5059c9c520ec8f3203648b5039dd367312e04462","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5059c9c520ec8f3203648b5039dd367312e04462"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 16:10:20.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/36d4989b6589954297f42694123a2cd6c0d275ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/36d4989b6589954297f42694123a2cd6c0d275ca"}]},"branch":"refs/heads/master"},"66cb0c6cccc7ed9e3c24d962868f9d75aec1b363":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-14 16:18:24.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/3"}}},"commit":{"parents":[{"commit":"29fb914af72eddc6db2e5701b9c8fde6cdc9b5d9","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/29fb914af72eddc6db2e5701b9c8fde6cdc9b5d9"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 16:18:17.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/66cb0c6cccc7ed9e3c24d962868f9d75aec1b363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/66cb0c6cccc7ed9e3c24d962868f9d75aec1b363"}]},"branch":"refs/heads/master"},"873d74f8faa990d1f5e08cf73597750ec0703d00":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-14 20:51:41.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/4"}}},"commit":{"parents":[{"commit":"ec36e70792184c526c0033e91d57dab5b4764a53","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ec36e70792184c526c0033e91d57dab5b4764a53"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 20:51:35.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/873d74f8faa990d1f5e08cf73597750ec0703d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/873d74f8faa990d1f5e08cf73597750ec0703d00"}]},"branch":"refs/heads/master"},"f27c77478718d99224936ae2b6fcf82c92e31cbf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-10-19 21:25:14.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/5"}}},"commit":{"parents":[{"commit":"4d382ea275b4d13910ce441ccb030ae5a1f59232","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d382ea275b4d13910ce441ccb030ae5a1f59232"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-19 21:25:07.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f27c77478718d99224936ae2b6fcf82c92e31cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f27c77478718d99224936ae2b6fcf82c92e31cbf"}]},"branch":"refs/heads/master"},"5d2b8fe60f90a2f713aaee444f88ac22cb74a894":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-10-24 11:29:39.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/6"}}},"commit":{"parents":[{"commit":"acc0e8178e4c756da13993bb9699625a78df8540","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/acc0e8178e4c756da13993bb9699625a78df8540"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-24 11:29:32.000000000","tz":120},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5d2b8fe60f90a2f713aaee444f88ac22cb74a894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5d2b8fe60f90a2f713aaee444f88ac22cb74a894"}]},"branch":"refs/heads/master"},"eb2943dfab00cb9e58eeddd97bc0d255f320120c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-11-25 17:43:56.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/85/684685/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/684685/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/684685/7"}}},"commit":{"parents":[{"commit":"9d8709930fdfcfdbc2270f18e5cca22128dd28e8","subject":"Fix defaults to storage and drive resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/9d8709930fdfcfdbc2270f18e5cca22128dd28e8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-09-25 10:13:06.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-25 17:43:50.000000000","tz":60},"subject":"Fix emulator socket binding options","message":"Fix emulator socket binding options\n\nPrior to this patch, dynamic emulator did not respect network\nendpoint address options passed through configuration file.\n\nWhile fixing this problem, file-based configuration code has been\ncleaned up and reworked to get loaded earlier.\n\nChange-Id: Id04caa4f1b0701af219cf85d6e30405ead097ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/eb2943dfab00cb9e58eeddd97bc0d255f320120c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/eb2943dfab00cb9e58eeddd97bc0d255f320120c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
