)]}'
{"id":"openstack%2Fsushy-tools~687539","triplet_id":"openstack%2Fsushy-tools~master~I1118e4a98278a517ae2a9c80873886a6bb6520f1","project":"openstack/sushy-tools","branch":"master","topic":"03-fix-storage-drives-defaults","hashtags":[],"change_id":"I1118e4a98278a517ae2a9c80873886a6bb6520f1","subject":"Fix defaults to storage and drive resources","status":"MERGED","created":"2019-10-09 11:29:06.000000000","updated":"2019-11-27 12:13:00.000000000","submitted":"2019-11-27 12:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687539-1574856683684-81677534","meta_rev_id":"7a7756d5f5803c97ad8cb821b8166d554470e15a","_number":687539,"virtual_id_number":687539,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 12:11:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-11-25 17:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-26 13:59:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 12:50:53.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-11-26 13:59:46.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-11-27 12:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f34cc5439054b3ef28b9ea15edf463ba60e17fbe","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-09 11:29:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4cd3566259dff08b7951d550331709d1e5f84fa","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-09 12:03:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02f963bb61121ab1e67ff95818712001ac311900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 12:45:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818c2e41cdbc4e4684f46bc1536c2186 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9ba0ca47ce94fea8b93973f2d47e3d7 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b21f3db3a00647e8ad8b1d0c023d770e : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2eed3c8b57544978f09f9d289fc0107 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35f2bcb2444e44819bce90067c09e900 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42b1f68c73464836a038f707051d9878 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6dbdea2b734b7fbd6eee60b3b2065e : SUCCESS in 4m 51s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/01364a659f1f4e50abe127df77b4bdfb : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"68233310796d17dd21abfba931180f6b4efa56c8","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 16:09:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"cf507cc86cab584122e2da9a34ee1f4dd6fe53ae","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 16:18:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b8c0ee11e5dfa0c4035a0e1ad916a0968c636241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:12:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391d89fdae654489b0e6ed0dcd9e444d : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b965ad1daa764d0e8b3f243323abf713 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4160986ea3401c98e9829a779aac9c : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c17a2a1fb3f848b4ba80b3f571ffd7ee : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b095c1f310a3427ebe7233c88b643abb : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c592375093334cc083ecb5f914e00a6a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d878c6b87d48448f157b6c6d77fb09 : SUCCESS in 6m 06s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9d0c3371b5994da48af0a3be1c42f6bc : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":4},{"id":"b22670f60e78a82c4f833832a6fc353393dde4cd","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-14 20:51:34.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"338ac2b6c7cb47274faa13cd49b7950558b101c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:50:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ad5e28757647b194b0b49f489175fe : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f15b3f97ff87479b892c37e736778217 : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261533a88a6a4db99c13a92d341dd7e8 : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a43aa27381ff469a8c14f1932b718061 : FAILURE in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e588fed5b544328b3c43cc0e6a5bc61 : FAILURE in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59832f50a7df432785c992ed23a23eb5 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1abcb52498403aa859eb05ba863aa8 : SUCCESS in 4m 55s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/cf73a1a03be64a1988f8a8d779b0371d : FAILURE in 54m 49s","accounts_in_message":[],"_revision_number":5},{"id":"4e2eee9bd64781719d9ec6319753bdea20bf4501","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-19 21:25:06.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"ccfc7e674bac4e1924dc25ab33a68a3d29d18cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 22:15:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f697c52d6c324f958eaeaa9054c63612 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b063a52ed5e04d4080a0018340583878 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27d78b34b78d4664af1bb9f52df32f59 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2743b412fcb446d7970c2e1e386a088d : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f544c6a473d04214a94e2951cdc9c7d4 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4da352906df482bb9b124f18174838a : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bb24e6280848cd85bb4bd5751a712c : SUCCESS in 5m 16s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0af4e8181c09426d8b09d8a88d9280aa : FAILURE in 46m 31s","accounts_in_message":[],"_revision_number":6},{"id":"44f4a4de14c4191936e7abc9f406967aa69227e7","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-24 11:29:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"091a6dff1265b51c73c686c8a24916cdaa5a714d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:38:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c4125367954f17976459e5c193b606 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ae069eb4bb749e0830c0d3eb8f23d6c : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75867ef784043d09e9bce0c628e4ef3 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39d989ece2da47ff8ec50f7080b6b8f4 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c89f7fe2f7554054893b885b56ab75fe : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fabdef33ef614ca58b11afbfbf49b5ab : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22dab123f0346fda0bf1e36ea0e2e24 : SUCCESS in 5m 33s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/38daae6626134c5da2127ff3a72e0916 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":7},{"id":"c2a9f6287019ae91528d18f6e413b6d72534c065","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-24 12:50:53.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"acb57809e68063b781a6c44dce6fc0e5e4844947","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-25 17:43:49.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"aebc2371b6bfd473d936ad113ea02ecfccc06a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:38:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4772979f7caa4d0b8905e2d486364c1a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fb0635bd44499bb00e3c985a1d08b4 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a5d978493494debbe60bbf14f4bbbde : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16587af255fa457eb0b5a895ae935431 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27221ce14d2e4e49a60d40f927eb262a : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff4887254564948adbcb5a8fe43dc92 : SUCCESS in 4m 50s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0414d601120b4e0ea899c7493efd8b05 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":8},{"id":"d6f4dda1cb2c9866ed2186da2a242537563b9c3b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-26 13:59:46.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"11ff57b71f9fd5368489a1803fc81349f75ed95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:59:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"794ddac579c2924be9bdb5b0d4d93ef40f376845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 14:53:23.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c30f213b844c4fa162f17e25c0b3e9 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40aa906c4f734e59b784b0ca7fab83d7 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c68b68a843aa4279a18718dd8276c329 : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a12aaf29217b4d828c82302bad22cb65 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab91ad6ebfc4a229126edbbd4f823f4 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72af09a604d948f7a230c1373b4c6457 : SUCCESS in 7m 32s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d707326155d140acb6b7dcc2afc26209 : FAILURE in 8m 24s","accounts_in_message":[],"_revision_number":8},{"id":"dc2ec464ba279920fae6e27929673b1b614f1d47","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-27 10:31:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8fab69983ec99eb71cf74643de4e3625b825888b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 11:21:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe9c19e8a8f468c8e2faca82368ead3 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbf71de7bfc94ce18fc3ffbefc1efd84 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4c00ac014d042848c9e295f4ca36bba : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ede66ab3f05147c6b2ee9bb4e45fcf08 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33f4158ee91845b6ad444a6b57a6e1be : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f5549b33494fb3b597b73622eba3ef : SUCCESS in 4m 50s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/37440b427a6548728e3b11a24738cfe1 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":8},{"id":"942ec17e1ff5f49d6101ad44709597e30ea349ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 11:21:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"1b5340470021260bcc0841be03afd1d3e39304fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"7c027ee07630532ece7516523af8df6b22028677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:11:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61285bda09f489c8d66c8af76092bae : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04085bd9a174e488d9ee30447ab203f : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1c46bcc9cdb403da1cf47d8210aff50 : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dca339a2749e4268ab024311f8aebacf : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1f445d4c04c48a5b3edff90bca081a7 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063c89fdf34c4173bd38b1da03a767fa : SUCCESS in 6m 36s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/7b9ea2d4055743318392dd0fd54cef0c : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":8},{"id":"17c3821180b7b605c05833b4bb3117210aee98f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:13:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9a21692d13144eea296b97dd4a441ed : SUCCESS in 1m 17s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649bc842122542e3b59306c4a232c3b6 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9d8709930fdfcfdbc2270f18e5cca22128dd28e8","revisions":{"844a7d019d022635f5f6d87ee243dd0c6bf53b6c":{"kind":"REWORK","_number":1,"created":"2019-10-09 11:29:06.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/1"}}},"commit":{"parents":[{"commit":"aed2f999c70ed28b4c755867e68bddd62bb73495","subject":"Merge \"Store virtual media device changes persistently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/aed2f999c70ed28b4c755867e68bddd62bb73495"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/844a7d019d022635f5f6d87ee243dd0c6bf53b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/844a7d019d022635f5f6d87ee243dd0c6bf53b6c"}]},"branch":"refs/heads/master"},"5e436052f17500ea5d29370a8362303c06744e04":{"kind":"REWORK","_number":2,"created":"2019-10-09 12:03:13.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/2"}}},"commit":{"parents":[{"commit":"aed2f999c70ed28b4c755867e68bddd62bb73495","subject":"Merge \"Store virtual media device changes persistently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/aed2f999c70ed28b4c755867e68bddd62bb73495"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 12:03:00.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5e436052f17500ea5d29370a8362303c06744e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5e436052f17500ea5d29370a8362303c06744e04"}]},"branch":"refs/heads/master"},"5059c9c520ec8f3203648b5039dd367312e04462":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-14 16:09:45.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/3"}}},"commit":{"parents":[{"commit":"2bf31058a75d8d17fd8aa5e0a02671af4c99f065","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2bf31058a75d8d17fd8aa5e0a02671af4c99f065"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 16:09:19.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5059c9c520ec8f3203648b5039dd367312e04462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5059c9c520ec8f3203648b5039dd367312e04462"}]},"branch":"refs/heads/master"},"29fb914af72eddc6db2e5701b9c8fde6cdc9b5d9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-14 16:18:16.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/4"}}},"commit":{"parents":[{"commit":"ec2a8dabc0ca496202c271327f859ebce74ad01f","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ec2a8dabc0ca496202c271327f859ebce74ad01f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 16:18:10.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/29fb914af72eddc6db2e5701b9c8fde6cdc9b5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/29fb914af72eddc6db2e5701b9c8fde6cdc9b5d9"}]},"branch":"refs/heads/master"},"ec36e70792184c526c0033e91d57dab5b4764a53":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-10-14 20:51:34.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/5"}}},"commit":{"parents":[{"commit":"090a3ec9c3e6c72fa86400f6653cd57f875af2ce","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/090a3ec9c3e6c72fa86400f6653cd57f875af2ce"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-14 20:51:27.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ec36e70792184c526c0033e91d57dab5b4764a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ec36e70792184c526c0033e91d57dab5b4764a53"}]},"branch":"refs/heads/master"},"4d382ea275b4d13910ce441ccb030ae5a1f59232":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-10-19 21:25:06.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/6"}}},"commit":{"parents":[{"commit":"fded6e373262a44b07965fa82dd397273720b75d","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fded6e373262a44b07965fa82dd397273720b75d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-19 21:25:00.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d382ea275b4d13910ce441ccb030ae5a1f59232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d382ea275b4d13910ce441ccb030ae5a1f59232"}]},"branch":"refs/heads/master"},"acc0e8178e4c756da13993bb9699625a78df8540":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-10-24 11:29:31.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/7"}}},"commit":{"parents":[{"commit":"8dec7774b141f85d9fdf0914b5ffb7c76d742e75","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8dec7774b141f85d9fdf0914b5ffb7c76d742e75"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-24 11:29:25.000000000","tz":120},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/acc0e8178e4c756da13993bb9699625a78df8540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/acc0e8178e4c756da13993bb9699625a78df8540"}]},"branch":"refs/heads/master"},"9d8709930fdfcfdbc2270f18e5cca22128dd28e8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-11-25 17:43:49.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/687539/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/687539/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/687539/8"}}},"commit":{"parents":[{"commit":"538d98dd1e0dccc540ae9dda2f226aaf8315224d","subject":"Fix incorrect parameter description in docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/538d98dd1e0dccc540ae9dda2f226aaf8315224d"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-10-09 11:28:33.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-25 17:43:43.000000000","tz":60},"subject":"Fix defaults to storage and drive resources","message":"Fix defaults to storage and drive resources\n\nIf storage and/or drive resource emulation is not set up\nvia sushy emulator configuration, some sane defaults must\nbe used. Otherwise the clients trying to read either of\nthese resources hit hard HTTP 500 response.\n\nChange-Id: I1118e4a98278a517ae2a9c80873886a6bb6520f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/9d8709930fdfcfdbc2270f18e5cca22128dd28e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/9d8709930fdfcfdbc2270f18e5cca22128dd28e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
