)]}'
{"id":"openstack%2Fsushy-tools~693639","triplet_id":"openstack%2Fsushy-tools~master~I9fdb038cc16147f9f7b248c48aa78a237cdba087","project":"openstack/sushy-tools","branch":"master","topic":"03-fix-log-via-flask","hashtags":[],"change_id":"I9fdb038cc16147f9f7b248c48aa78a237cdba087","subject":"Always log via Flask application","status":"MERGED","created":"2019-11-10 22:33:07.000000000","updated":"2020-03-11 11:27:38.000000000","submitted":"2020-03-11 11:25:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693639-1583925956634-660eed34","meta_rev_id":"e68ad94d0e56bc41849ef6e26e67ceeeac65298f","_number":693639,"virtual_id_number":693639,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 11:25:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-03-11 10:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2020-03-10 11:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-03-11 10:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 01:35:15.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-12-03 22:38:45.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-03-10 11:20:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-11 10:30:15.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-11 11:25:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc1c80553e1a75f56f0c9d06de94d4bf13a52b11","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-10 22:33:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7259e05174b212de743b9d2658dd9c2fd59c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-10 23:33:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3996d1a75d2440b49f6d9e7368d3aa8b : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46f20211c91f4cb18da63bcd10c52161 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d85f11b3e1384f6bb9db64674d52f89a : SUCCESS in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f160e4ad63c44cf7b453967387499e33 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c534d4254b334b10b3f2995888eb8482 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec74c918217b413791bb75b03eedc302 : SUCCESS in 4m 30s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/2662aa7c003840bd830366d295526208 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ee76bfae21b4eacc892b360f338e18b3f316c4b9","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-11-25 06:11:44.000000000","message":"Patch Set 1:\n\nHi Ilya, I think we don\u0027t need to embed flask.logger into base classes, this makes things complicate. Can we create a logging module initialized with the flask instance at start up, and use it onwards?","accounts_in_message":[],"_revision_number":1},{"id":"14791ad5b6219a2e2b4640d294a01a4ccbd950c0","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-25 17:44:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"74cead910e84adb786ea50dc2db955a5d0ca3b57","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-25 17:59:19.000000000","message":"Patch Set 2:\n\n\u003e  I think we don\u0027t need to embed flask.logger into base classes, this makes things complicate. Can we create a logging module initialized with the flask instance at start up, and use it onwards?\n\nI am not sure...\n\nLook, we still need to pass Flask configuration to resource classes (some resources are configurable). Configuration is associated with Flask app object. Previously we just picked Flask configuration dict and passed it resource classes.\n\nThen I discovered, that it\u0027s better to log through Flask.Logging object is also associated with Flask app. So I thought that something else might come up going forward. To make it future-proof, I tried to pass the entire Flask app object to each resource. Does it make sense, is it worth it?\n\nIf you think we are better off staying with just config being passed to resource classes, then I think global logging object makes perfect sense.\n\nHowever, if we gonna give resources the entire Flask app object, then may be calling its logger would be overall easier and more consistent.\n\nWDYT?","accounts_in_message":[],"_revision_number":2},{"id":"f4de4bacb1c1596f96cc80c94f5de6485b8ab838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:51:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8486910455324666a5121f152b44b1e9 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a410e6aab9486a92c286c0122b4d66 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5502feeb37c44c2981767c626c5a1bcd : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26bdfeaa5a654bf09a359716ad9e69ae : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d88aa9a376814c9ea4ffc5330641ce95 : SUCCESS in 5m 38s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/a614220dd6594ad1b4c985da6ab81a70 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":2},{"id":"52e1f89003f2df2f46045823310fd6919fadc337","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-11-26 01:35:15.000000000","message":"Patch Set 2:\n\nHi Ilya, the flask logger is assigned to the class, actually this is not much different with a global variable, so I think it might be more convenient to put it into a single module, unless we always do logging in methods. You are far more familiar with this project than me, so I am just thinking aloud and raising a proposal.","accounts_in_message":[],"_revision_number":2},{"id":"5a1c04fba0020fbc8602782c142231aabb232edd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-26 14:04:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"41b869abfc3d0cdc059ae3d0ca098291a94dec02","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-26 14:29:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ed14793927b7b85f0f394ecaa0004c26f39afad","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-26 14:39:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3cc18a8fa88ae850066b9dfbd97f5905d18be33","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-26 14:44:51.000000000","message":"Patch Set 2:\n\n\u003e Drivers calling back to the API layer seems a big violation of layers.\n\nWith Flask, it does not appear like a pure API layer, but rather a resource management entity.\n\nIf it was only the API part, then it would not probably own logging and configuration - these things are generic enough to be utilized by many parts of the application, not just to the API layer.","accounts_in_message":[],"_revision_number":2},{"id":"858e5b8ab7b03a8f8e9b0bd9b506a4df4f767657","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-27 13:08:21.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b77a9a64c31c74009cc82ce2dec9fe262eec3937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:00:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591cdc446d9d4f3ea1170c0c8c430094 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4689d4ed76a43d597ea5d1e59592358 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1874ca11d31f4167b6b90c2c5a0c2659 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/116e792463b94fcbb3cc53d4f25c3022 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cbfce867c9647ab9b69e8a0648c4ff6 : SUCCESS in 10m 08s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/77e0b80891b042f2b010228c8665321c : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":3},{"id":"99924fff26a5d584004c237072d6b8b0b2d1efc4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-28 18:46:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4c7ddf2d602714750f80675d8f70994fb96484bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 19:36:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e2aa524e3a4b7fbee469a6744a70bb : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf49d13d7e0b4554b771f6286e157b1b : FAILURE in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fdb2cc5f9c64b72b6f1d3e7563b0b66 : FAILURE in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9121ea6fc9fa4a2f8ef92d049b5304d8 : FAILURE in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7850401c8f047508409419c8ac5615c : SUCCESS in 6m 05s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/587eca0dc07e49319c6eafad01ad23f0 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6526e75f37601344e420de5730ded5fe2525bfa0","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-29 17:45:26.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4e190d939d44651cde5f4076a9c8a551f060b1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 18:39:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41529b17961f4f8ba8636841fdd55628 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6447818e88c84a4b81ad3be3397e5457 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62cc259ba56044b5b04d5d8036885528 : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b28850ed547e4314ac9670283bcf045f : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a2e7eeabf44cf2bfe3c6555a866082 : SUCCESS in 4m 51s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0587b90bd5a94875a9f4fa42c4d1e2af : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":5},{"id":"4753419948ca3ed0367ad2227ad78ff594aea896","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-12-02 12:40:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"afc4849f04874292c00ecf5600b209db8d68247d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:52:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aadbe0118b34cf8b433412b0adb46b2 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9d689e895ff447fb1b0e422793a95a8 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd66024226754b5e8f059d7b579c4875 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec399411f8154b07a5019d77d5d20865 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdacc32e0f354f2a9707e9365abe1f73 : SUCCESS in 6m 17s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/162845074acb4dba88ef2f92095b4987 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":6},{"id":"44b3b0b50b781af2bf2184efea61e30a99a1a712","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-12-03 22:38:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nI concur with Dmitry. Adding -1.","accounts_in_message":[],"_revision_number":6},{"id":"1da22e801c201414bd52132cf9603dd600eb656f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-03 12:54:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dba1880fcbfc2dbdeea20dc335f584ed4fda820a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-03 13:06:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7d5003431f97bfb841589268f57e0d0bd1cd50ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 13:27:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dade3db821e448d590c0b52e3006aa13 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e30efdee65d449548885bfab61eb72ad : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7167661ad6394087bbed340baa42418c : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d8fb672b9c489bb982103e267fc2ce : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dce1c735e964ad99cc78b98836025c9 : SUCCESS in 6m 03s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/6df3f88951a347dfb947ba3f48a62ab3 : FAILURE in 31m 16s","accounts_in_message":[],"_revision_number":7},{"id":"23263743f7f819572725a641ae817e0d885997c1","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-03 13:34:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9d2c7e60fa3c25753c297244a0a86340f17ed751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 14:35:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f16547950ccd499994b9d413449ab829 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8089f180b314e5e839845e75d6cc248 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9db96e5ae984fc09c89bd77cfb77a96 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad183cb831354a8382385ff49f591028 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a758edd9e04f47a27052876e5cce40 : SUCCESS in 4m 58s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/76d067ba42384916a5a15093f2aa0242 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":8},{"id":"70d60b2ee7acdd50b43171b848511fb91bc3dcc7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-10 11:20:07.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9dd91cd3ed59670282960a35c7022f91cf75d30d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-11 10:30:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"445b1f878df58bda5e99609f4f2194a5328d9e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 10:30:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6433c3c51d42292262462b69e4af0111d1f11c77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:25:56.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76eb65ed63ab41cfaab50427a61288fe : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/696f411629d343b29053500f6c269040 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab3f4899840c490080ef7a3917c2fc7d : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9290276598449987cdbd113d8f8026 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c2c002044e443e8370bedc0121a267 : SUCCESS in 6m 47s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/dca58b75225040e685559d6c296222ff : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":8},{"id":"12a65f6752ca3664b6c1da2c5588de76b98fc319","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:25:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"c3ec4caa347321457ed82641ff3a4b54591162c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:27:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e203e2dbb2684ca4b91444b979b16fbd : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"663559c7e8befb581327acd2439534acb3521453","revisions":{"3c7e9a5e39cfd0b1ccef26cf76812a25f5785864":{"kind":"REWORK","_number":1,"created":"2019-11-10 22:33:07.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/1"}}},"commit":{"parents":[{"commit":"71d4562acc4fdc7be56d3bca37b86d7f9906a4c9","subject":"Fix `BootSourceOverrideMode` enumeration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/71d4562acc4fdc7be56d3bca37b86d7f9906a4c9"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:32:45.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3c7e9a5e39cfd0b1ccef26cf76812a25f5785864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3c7e9a5e39cfd0b1ccef26cf76812a25f5785864"}]},"branch":"refs/heads/master"},"30df4770df402b90b244e5e15d227b449b889268":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-25 17:44:12.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/2"}}},"commit":{"parents":[{"commit":"4279ee0622db5654cbbb7479cbcf21747b29780c","subject":"Fix `BootSourceOverrideMode` enumeration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4279ee0622db5654cbbb7479cbcf21747b29780c"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-25 17:44:03.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/30df4770df402b90b244e5e15d227b449b889268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/30df4770df402b90b244e5e15d227b449b889268"}]},"branch":"refs/heads/master"},"bd96ac9db4d09980e53bb9e2948ca624d9165866":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-27 13:08:21.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/3"}}},"commit":{"parents":[{"commit":"723ffd231fa4269af8a4136cbea8cb1d661b2ffd","subject":"Fix libvirt CD image insertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/723ffd231fa4269af8a4136cbea8cb1d661b2ffd"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-27 13:07:54.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/bd96ac9db4d09980e53bb9e2948ca624d9165866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/bd96ac9db4d09980e53bb9e2948ca624d9165866"}]},"branch":"refs/heads/master"},"075ab280007ccf70c96482d90574d3bb5fa95576":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-28 18:46:45.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/4"}}},"commit":{"parents":[{"commit":"db91e0c15de9a45eb1e94fb06097096482a508a8","subject":"[WIP] Bring libvirt domain down prior to any change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/db91e0c15de9a45eb1e94fb06097096482a508a8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-28 18:46:12.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/075ab280007ccf70c96482d90574d3bb5fa95576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/075ab280007ccf70c96482d90574d3bb5fa95576"}]},"branch":"refs/heads/master"},"1f15b615eae361ddb3fd7e61e01bec84a8182fe9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-29 17:45:26.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/5"}}},"commit":{"parents":[{"commit":"484699813b8e88ff62e89192849e1f0177283099","subject":"Retain boot device selection upon boot image change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/484699813b8e88ff62e89192849e1f0177283099"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-29 17:45:20.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1f15b615eae361ddb3fd7e61e01bec84a8182fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1f15b615eae361ddb3fd7e61e01bec84a8182fe9"}]},"branch":"refs/heads/master"},"24cd736b687dc9d4ae824b951f52fde6d6185cd6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-02 12:40:39.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/6"}}},"commit":{"parents":[{"commit":"907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab","subject":"Retain boot device selection upon boot image change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-12-02 12:39:46.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/24cd736b687dc9d4ae824b951f52fde6d6185cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/24cd736b687dc9d4ae824b951f52fde6d6185cd6"}]},"branch":"refs/heads/master"},"a7b28467fe2af2e1850df0c721361c57eaf10dc4":{"kind":"REWORK","_number":7,"created":"2020-01-03 12:54:05.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/7"}}},"commit":{"parents":[{"commit":"907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab","subject":"Retain boot device selection upon boot image change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-03 12:53:25.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a7b28467fe2af2e1850df0c721361c57eaf10dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a7b28467fe2af2e1850df0c721361c57eaf10dc4"}]},"branch":"refs/heads/master"},"663559c7e8befb581327acd2439534acb3521453":{"kind":"REWORK","_number":8,"created":"2020-01-03 13:34:01.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/39/693639/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/39/693639/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/39/693639/8"}}},"commit":{"parents":[{"commit":"907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab","subject":"Retain boot device selection upon boot image change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/907c3ad6d50ef7fc11bfb7f324c91f30016ba2ab"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-11-10 22:31:03.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-03 13:33:28.000000000","tz":60},"subject":"Always log via Flask application","message":"Always log via Flask application\n\nPrior to this patch, sushy-emulator used two ways of logging: via\nFlask app object and via Python logging module. This arrangement\nseems to be leading to missing log messages in the CI.\n\nThis patch makes sushy-emulator consistently logging via Flask\napp object.\n\nChange-Id: I9fdb038cc16147f9f7b248c48aa78a237cdba087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/663559c7e8befb581327acd2439534acb3521453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/663559c7e8befb581327acd2439534acb3521453"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
