)]}'
{"id":"openstack%2Fsushy-tools~708495","triplet_id":"openstack%2Fsushy-tools~master~Idfc63ae509fefdb1436b1a70156a86a9b2d3a677","project":"openstack/sushy-tools","branch":"master","topic":"708495","hashtags":[],"change_id":"Idfc63ae509fefdb1436b1a70156a86a9b2d3a677","subject":"Fix attribute error in `set_boot_image()`","status":"ABANDONED","created":"2020-02-18 19:22:47.000000000","updated":"2020-07-30 12:35:54.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a61033ad4ed1742f933fdb3d77c3c6d6dc6067b","_number":708495,"virtual_id_number":708495,"owner":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-28 09:05:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 10:48:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-13 09:50:51.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2020-07-28 09:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"560fe9b367e9b749ce6fac3e83fbb7856450ccbc","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2020-02-18 19:22:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"101b97a7927e16b5fb9cc2b4b592c496d602d5e6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-02-18 19:49:31.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nReno would not hurt as well. People may be looking for fix information there.","accounts_in_message":[],"_revision_number":1},{"id":"2a54faff63be4c004d8a7edb2ddb300f4bd05016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:09:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8402bca5364ab3b69c68ffaefb16fb : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f5f9d0d31df4eb1a51aeea85556353d : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da498b79194e4815a05a23869050f44a : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f13b0a2cbade43eba55ee5139d644284 : FAILURE in 8m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb27fc83bf7a4a8baf110cbe618b02c1 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c615feb6bd93415ca0fe597a6d7fac5a : SUCCESS in 6m 55s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/1772c6e0006d4affafe9dd95784af07b : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a5c93513fc953f30d0ee671f483be8cfa9a92ef3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-19 09:10:16.000000000","message":"Patch Set 1:\n\nEven if it\u0027s still non-voting, we should keep an eye on the Python 3.8 test and make sure it passes.\nI also agree a release note would help.","accounts_in_message":[],"_revision_number":1},{"id":"e1071dd75252879c47f623c6c466e69bd8ed209f","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2020-02-23 23:31:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33e6fb910806b3abb4191afc1fcb0148ee42aead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 23:53:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ecf9ad8f8c40cdae94e11a07606935 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaede5645ee6436f8d182a013f7bb643 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b406a216cf64309b835a12e77eb418b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a9d51c53f3842faabb1a2766c08520b : FAILURE in 4m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aa7172c69c242e4b61ccb27f5dba51c : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd661901c8bc4c95818ac61dc1231fc5 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c335e4610fa476c88fee1f0380168ef : SUCCESS in 6m 45s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/144db76cea4e42f2af832ef422990ef2 : FAILURE in 16m 13s","accounts_in_message":[],"_revision_number":2},{"id":"677efdc98dd23ab94e01d530e44b25c3690bbd1a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-27 10:48:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1edb2bf10c7c9331096044937f8706513922f236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:42:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7444243f30e414e940ba4b4e5c2ec21 : SUCCESS in 14m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/278d30baeed84c33b69f649a35cb84ce : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/870345e760ad4587b5ba00ecead52dc9 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dde21f63f2b24eb188624967faf7914f : FAILURE in 11m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4d4456d127b4c4bb35a15ad8089a77d : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87c8151747b84c968ee95cd8a52620d9 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f622cc8afe4c40d780a1f7eaf99ccb90 : SUCCESS in 26m 32s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/f0c503f96d31448ba57df795364c45c8 : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a8b5dfcdde47a7616d76c48dd1ee56e72084db59","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-02-27 17:02:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nBTW, a possible bonus patch (separate from this one) could be to fix 3.8 test failure:\n\n    File \"/home/zuul/src/opendev.org/openstack/sushy-tools/sushy_tools/tests/unit/emulator/resources/systems/test_libvirt.py\", line 306, in test_set_boot_device_network\n            self.assertIn(expected, conn_mock.defineXML.call_args[0][0])","accounts_in_message":[],"_revision_number":2},{"id":"16d76621a768fb9fbc63d517510387abfda157c7","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2020-03-10 20:52:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0f3b64e47af5c288d2df775e16b9f5e0591e045","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2020-03-10 20:53:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e554cdd9bd2dee6dde06e40a90affeddc318d5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:37:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8431168e6a654bd0a292d736b5af1432 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a37da554f89340c6a4a676cb2692b599 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e31fbca19c64fffa2583a0a7012b3a5 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8005e7c689cf4a449bf67731c38373ac : FAILURE in 4m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/555a094cd3464c6ebcfcce274aef167a : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8654df5c660409683517110fff8b518 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59905264fc8b46339d5777850ba727c4 : SUCCESS in 6m 27s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/ab0e548c0bb245b9a2206257e725246e : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d5fe92a1936e42cd4f073c6d790b022b26e3a282","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-03-13 09:50:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f778662a7415559c0671ed182266d02688b7a5a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-06 13:59:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8f260914ae5b427df2023b1fb38742e013d6662e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:53:28.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a498c54adae443519e6561b9776492f5 : FAILURE in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2bc274ac954f10944b1f6aba1785be : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7cd56c1db084bb0ae6032799065f970 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6935d7770b64be09b938150f5e7720d : FAILURE in 6m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a8622a042dd4779afa854b31acdf76a : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f44419defef4f5f851a22b6a7730c44 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009c5d090384437fac873efe92c29f22 : SUCCESS in 7m 32s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/ca1e48842b45414e94c644ac0154a72b : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":4},{"id":"86098a5f417980f0e2b5e693b3b6010013eb993d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-28 07:28:56.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d64493152a15f0b8695b5f86d16b4cadc088993d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-28 07:32:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e0b5a03094b771d76ebb4b7159191f07d6f4e6a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-28 07:34:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a8c411854bcb63f8726418367178cb3f6a21915a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 09:05:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e066efad7144c7b8332d3be1bb7d9ae : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b92bf120c844fd3b04a7d65ed1a56e7 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb069e478df046fe8a131a75947fd5be : SUCCESS in 6m 40s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/249ae8af9a7f401cb7d6a809dc13f02c : SUCCESS in 1h 27m 29s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e306a22b9ce24325a628787997acf574 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":7},{"id":"c212277e10e760c2c733dd90e3e3660d4c33545d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-30 12:35:54.000000000","message":"Abandoned\n\nalready fixed","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"afc463797625b84e8b6f75b42a1a0f16917933dc","revisions":{"14308c5e2b0f13b898aa7f3500f0d810319d04ee":{"kind":"REWORK","_number":1,"created":"2020-02-18 19:22:47.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/95/708495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/1"}}},"commit":{"parents":[{"commit":"c08a57cc5addd32e37dccc7e19791181a9fc68d3","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c08a57cc5addd32e37dccc7e19791181a9fc68d3"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:22:45.000000000","tz":60},"subject":"Change logger call","message":"Change logger call\n\nlogger would be call without self\n\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/14308c5e2b0f13b898aa7f3500f0d810319d04ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/14308c5e2b0f13b898aa7f3500f0d810319d04ee"}]},"branch":"refs/heads/master"},"d0de568e45c7696ea223cc44e27c4d057c1c5dfe":{"kind":"REWORK","_number":2,"created":"2020-02-23 23:31:36.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/95/708495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/2"}}},"commit":{"parents":[{"commit":"c08a57cc5addd32e37dccc7e19791181a9fc68d3","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c08a57cc5addd32e37dccc7e19791181a9fc68d3"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-23 23:30:45.000000000","tz":60},"subject":"Error on inserting virtual media device","message":"Error on inserting virtual media device\n\ngetting libvirtdriver does not have logger attribute ERROR while using\nlibvirt emulator to insert a virtual media device.\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d0de568e45c7696ea223cc44e27c4d057c1c5dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d0de568e45c7696ea223cc44e27c4d057c1c5dfe"}]},"branch":"refs/heads/master"},"f0c3fef7a8ca1078c18af130aae1b96354f12bd3":{"kind":"REWORK","_number":3,"created":"2020-03-10 20:52:45.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/95/708495/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/3"}}},"commit":{"parents":[{"commit":"c08a57cc5addd32e37dccc7e19791181a9fc68d3","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c08a57cc5addd32e37dccc7e19791181a9fc68d3"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-03-10 20:50:35.000000000","tz":60},"subject":"Fix attribute \u0027logger\u0027 reference in set_boot_image exception method","message":"Fix attribute \u0027logger\u0027 reference in set_boot_image exception method\n\nInserting a virtual media device failed invoked the set_boot_image\nmethod exception in return, instead of having the message error, we got\n\"logger\" attribute not found! which is due to an incorrect reference;\n\"self._logger\" instead of just \"logger\".\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f0c3fef7a8ca1078c18af130aae1b96354f12bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f0c3fef7a8ca1078c18af130aae1b96354f12bd3"}]},"branch":"refs/heads/master"},"0eb626cf78f7fbaf9975285f27277badd8e2a318":{"kind":"REWORK","_number":4,"created":"2020-04-06 13:59:06.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/95/708495/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/4"}}},"commit":{"parents":[{"commit":"c08a57cc5addd32e37dccc7e19791181a9fc68d3","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c08a57cc5addd32e37dccc7e19791181a9fc68d3"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-06 13:55:32.000000000","tz":120},"subject":"Fix attribute error in `set_boot_image()`","message":"Fix attribute error in `set_boot_image()`\n\nFixes a bug in attribute reference that occurs on\nvirtual media insert operation logging.\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0eb626cf78f7fbaf9975285f27277badd8e2a318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0eb626cf78f7fbaf9975285f27277badd8e2a318"}]},"branch":"refs/heads/master"},"2f94f97d26f71609aa24bcfd53444cd5b7296189":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-28 07:28:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/708495/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/5"}}},"commit":{"parents":[{"commit":"ffba0c7bcd0b7b301f65e3083eb57636e16cee34","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ffba0c7bcd0b7b301f65e3083eb57636e16cee34"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-28 07:28:49.000000000","tz":120},"subject":"Fix attribute error in `set_boot_image()`","message":"Fix attribute error in `set_boot_image()`\n\nFixes a bug in attribute reference that occurs on\nvirtual media insert operation logging.\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2f94f97d26f71609aa24bcfd53444cd5b7296189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2f94f97d26f71609aa24bcfd53444cd5b7296189"}]},"branch":"refs/heads/master"},"d67516fe9194b584ea3f946c11157f89570ac585":{"kind":"REWORK","_number":6,"created":"2020-07-28 07:32:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/708495/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/6"}}},"commit":{"parents":[{"commit":"ffba0c7bcd0b7b301f65e3083eb57636e16cee34","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ffba0c7bcd0b7b301f65e3083eb57636e16cee34"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-28 07:32:49.000000000","tz":120},"subject":"Fix attribute error in `set_boot_image()`","message":"Fix attribute error in `set_boot_image()`\n\nFixes a bug in attribute reference that occurs on\nvirtual media insert operation logging.\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d67516fe9194b584ea3f946c11157f89570ac585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d67516fe9194b584ea3f946c11157f89570ac585"}]},"branch":"refs/heads/master"},"afc463797625b84e8b6f75b42a1a0f16917933dc":{"kind":"REWORK","_number":7,"created":"2020-07-28 07:34:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/708495/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/708495/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/708495/7"}}},"commit":{"parents":[{"commit":"ffba0c7bcd0b7b301f65e3083eb57636e16cee34","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ffba0c7bcd0b7b301f65e3083eb57636e16cee34"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2020-02-18 19:10:23.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-07-28 07:34:13.000000000","tz":120},"subject":"Fix attribute error in `set_boot_image()`","message":"Fix attribute error in `set_boot_image()`\n\nFixes a bug in attribute reference that occurs on\nvirtual media insert operation logging.\n\nStory: 2007329\nTask: 38851\nChange-Id: Idfc63ae509fefdb1436b1a70156a86a9b2d3a677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/afc463797625b84e8b6f75b42a1a0f16917933dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/afc463797625b84e8b6f75b42a1a0f16917933dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
